)]}'
{"id":"openstack%2Fansible-role-systemd_mount~776962","triplet_id":"openstack%2Fansible-role-systemd_mount~master~I5a83a7f237e4b7d86e8d7ee08b0e8c6b4f8a44f8","project":"openstack/ansible-role-systemd_mount","branch":"master","topic":"osa-facts","hashtags":[],"change_id":"I5a83a7f237e4b7d86e8d7ee08b0e8c6b4f8a44f8","subject":"Use ansible_facts[] instead of fact variables","status":"MERGED","created":"2021-02-22 16:56:25.000000000","updated":"2021-02-24 11:22:18.000000000","submitted":"2021-02-24 11:20:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776962-osa-facts","meta_rev_id":"7a9583f289f5865fb6cb568520ed63f4fb726ddb","_number":776962,"virtual_id_number":776962,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-24 11:20:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-02-24 09:31:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-02-23 14:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-02-24 09:31:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 18:53:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 14:15:25.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-24 09:31:37.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"f24c49760dfd593de799700167fc2c0f463c804e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 16:56:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e64a1ced47db8b51df0871c535d45f5e0d224205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 18:53:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cc1694d8331743aeabacbd3f91cec4b7 : FAILURE in 7m 43s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/ddb4039418b44de7ac64bcaefdf7e6ee : FAILURE in 7m 46s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/dfee140cd28c4e60a57c67f8951648cf : FAILURE in 9m 08s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ad43310a3e8412ebb056e7f67532817 : FAILURE in 12m 06s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00f91ee956994cf9b25aeced843d3fb7 : FAILURE in 12m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6990230ad5284dcfa1d8a72cca2b5857 : RETRY_LIMIT in 15m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/12c2a74647e94f2e8253c83287bd4872 : RETRY_LIMIT in 19m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9bca51af5f894f13ab98fe2999fc7697 : RETRY_LIMIT in 10m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9266a5821878419e804670de1e0585a1 : RETRY_LIMIT in 19m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/83fadff88ca5448ba5f29eb46fc2cf89 : SUCCESS in 1h 45m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86eee6b2542d4ccfb6cb0fa6c2edce7d : SUCCESS in 1h 28m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7d149b1c865e4ae19081155a073dfe5d : RETRY_LIMIT in 8m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcb2a0f289294687bf28077a16fd98c9 : RETRY_LIMIT in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2dcab6fd984e78a6302f5ca03aea8f : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"08b1fd4366ef1cabe9d8f4acc2efa5b964e5886e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 23:06:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0b26b19d6e40ade9e3a81be6f691199f240c12f2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-22 23:15:06.000000000","message":"Topic set to osa-facts","accounts_in_message":[],"_revision_number":2},{"id":"64ce80494524d15d78fd650585d0822101af5be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 00:58:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5ad8e56e71194efeb144920a6d55d7bb : SUCCESS in 8m 40s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/cc86b652e7b64c9f84d2aa824b5f70e3 : SUCCESS in 8m 03s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/dc36d6b94b8843859c520cb57cfd31b3 : SUCCESS in 10m 14s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05c54431ceb447aa9ae08dfa02d656f8 : SUCCESS in 8m 22s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0befe8f8d05c4c499791d0a3db4c6e51 : SUCCESS in 9m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/69816d6838ae4d67b02d31f13390a479 : SUCCESS in 1h 06m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/150fa2d6bbbc43eda19e3b0025bb5cd4 : SUCCESS in 1h 14m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00ef3219481e42f9ab5cea5b504e462b : SUCCESS in 1h 20m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ae582ad71194ce18bcdec1eded540b7 : SUCCESS in 1h 01m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bdd1ba47fe724c06896915977a842325 : SUCCESS in 1h 45m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15212048655f483684fb79e8cc808114 : SUCCESS in 1h 30m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8e0b0bb18b144ef39cf5c5d61712cb28 : FAILURE in 35m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d23f15a4921e4e9d84c1d0c580744308 : SUCCESS in 58m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af7df120d174486b2214855ae2dbb25 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9c70c4c28034ad6b3efed4b95821f074266e7865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-23 09:25:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1a9530300e402a161e3f3951fb7558e3762a09c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 11:27:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0df25f0dc96c48ba99aeede15cdd83e1 : SUCCESS in 9m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/a6c3d97eeaad4d60914cfcbb00269bdd : SUCCESS in 8m 36s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/0581988f54234f48b18daba1f951a376 : SUCCESS in 14m 10s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bea5d08bd3394d78afbd9a433dbf633f : SUCCESS in 9m 13s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3753896b899843edb9e5b1e2aa5bb872 : SUCCESS in 11m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f1c26ab68cb847018a49e6c8fd21d002 : SUCCESS in 1h 02m 47s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a71a854eaceb46e4b2e2fef953f549b8 : SUCCESS in 1h 08m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b065c1b5c6c42458392c0a85b76d02b : SUCCESS in 1h 07m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/744fbe2c32f742c0b25feda1e72344a0 : SUCCESS in 1h 04m 57s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c062843772354158a0e17f647e064520 : SUCCESS in 1h 48m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4af41da9d974909b88cc9c88e709b49 : SUCCESS in 1h 42m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d3721548e13a43a4a162b805cd0b5d4d : FAILURE in 37m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b1c8f4eef9c46b8979aa126165e27e4 : SUCCESS in 51m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ee2a02f728489eb1beff420579781e : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":3},{"id":"6dd18a31a5936fdd4876179bd8b83b64703cfa5f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-23 14:15:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"867250e7d547f04e77b37133d465ff8521f4dff9","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-24 09:31:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"28d2f90e43a6172612863580a9666c8dfa754157","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 09:31:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ae5328ffaff516fb5f6a0ada7b4a32a89011a703","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 11:20:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1720b3d9b73242139eb3a8d639025858 : SUCCESS in 10m 11s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/4893f70544af4a608585cdb1b28ddf72 : SUCCESS in 9m 00s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8924c46c5cb04bc8b6d5be0ad544c51f : SUCCESS in 10m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0825fb88c8f94298a78e0956d914e9fa : SUCCESS in 1h 41m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6920f852947349c78fa4db04261022d2 : SUCCESS in 1h 10m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20d6529bc91d457c92648559e239a6fc : SUCCESS in 1h 26m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3858ca2485614cf2b282f88372d4961a : SUCCESS in 1h 17m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5ccda160c9e942a6bec1989156fb2c22 : FAILURE in 34m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dde8f06223c54645a325aff50206e442 : SUCCESS in 51m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202294362a6b457e9b11806d712df6c0 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":3},{"id":"753b992c1834e4fde60b860d68fa7c58094b17a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 11:20:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7a9583f289f5865fb6cb568520ed63f4fb726ddb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 11:22:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e7244ab0a7947a8b3da5ec1fee32970 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e5cc6f8990e82b92bc489c701f849120ec23f2fa","revisions":{"138cd8a491f4083856efbcfbffe7206050e337f3":{"kind":"REWORK","_number":1,"created":"2021-02-22 16:56:25.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/62/776962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/62/776962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/1"}}},"commit":{"parents":[{"commit":"483b10e5e989692f358bba49f417be0fbb6e00d8","subject":"Add option to use s3fs with role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/483b10e5e989692f358bba49f417be0fbb6e00d8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-22 16:56:05.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-22 16:56:05.000000000","tz":0},"subject":"Use ansible_facts[] instead of fact variables","message":"Use ansible_facts[] instead of fact variables\n\nSee https://github.com/ansible/ansible/issues/73654\n\nChange-Id: I5a83a7f237e4b7d86e8d7ee08b0e8c6b4f8a44f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/138cd8a491f4083856efbcfbffe7206050e337f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/138cd8a491f4083856efbcfbffe7206050e337f3"}]},"branch":"refs/heads/master"},"16fe92d82801eb63c31e083bb3abf80c3ab7324b":{"kind":"REWORK","_number":2,"created":"2021-02-22 23:06:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/62/776962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/62/776962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/2"}}},"commit":{"parents":[{"commit":"483b10e5e989692f358bba49f417be0fbb6e00d8","subject":"Add option to use s3fs with role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/483b10e5e989692f358bba49f417be0fbb6e00d8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-22 16:56:05.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-22 23:05:58.000000000","tz":0},"subject":"Use ansible_facts[] instead of fact variables","message":"Use ansible_facts[] instead of fact variables\n\nSee https://github.com/ansible/ansible/issues/73654\n\nChange-Id: I5a83a7f237e4b7d86e8d7ee08b0e8c6b4f8a44f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/16fe92d82801eb63c31e083bb3abf80c3ab7324b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/16fe92d82801eb63c31e083bb3abf80c3ab7324b"}]},"branch":"refs/heads/master"},"e5cc6f8990e82b92bc489c701f849120ec23f2fa":{"kind":"REWORK","_number":3,"created":"2021-02-23 09:25:58.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/62/776962/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/62/776962/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/62/776962/3"}}},"commit":{"parents":[{"commit":"483b10e5e989692f358bba49f417be0fbb6e00d8","subject":"Add option to use s3fs with role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/483b10e5e989692f358bba49f417be0fbb6e00d8"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-22 16:56:05.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-23 09:25:49.000000000","tz":0},"subject":"Use ansible_facts[] instead of fact variables","message":"Use ansible_facts[] instead of fact variables\n\nSee https://github.com/ansible/ansible/issues/73654\n\nChange-Id: I5a83a7f237e4b7d86e8d7ee08b0e8c6b4f8a44f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/e5cc6f8990e82b92bc489c701f849120ec23f2fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/e5cc6f8990e82b92bc489c701f849120ec23f2fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
