)]}'
{"id":"openstack%2Fansible-role-systemd_mount~780745","triplet_id":"openstack%2Fansible-role-systemd_mount~master~I55037805e59cb6d727593f467e98568a4c030e74","project":"openstack/ansible-role-systemd_mount","branch":"master","hashtags":[],"change_id":"I55037805e59cb6d727593f467e98568a4c030e74","subject":"Remove references to unsupported operating systems","status":"MERGED","created":"2021-03-16 08:57:52.000000000","updated":"2021-03-16 16:58:07.000000000","submitted":"2021-03-16 16:55:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780745","meta_rev_id":"7bf9811bb13ed68592c1c7b4b86a9ef5474d6daf","_number":780745,"virtual_id_number":780745,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-16 16:55:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-03-16 14:15:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2021-03-16 15:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2021-03-16 15:11:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 10:52:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-16 14:15:43.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2021-03-16 15:11:51.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"616c17c1ab6be2dbd109d65068b430e711f0418a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-16 08:57:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89a45074a783c4c74bd4dec744d73105aa0028ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 10:52:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/95c23ce2de524c70a001e8a7bd9b755e : SUCCESS in 14m 40s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/8de361113cad4d88a48be8ecb3c7748f : SUCCESS in 8m 41s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/b709875b8a194b10b6ea4cba142a02e9 : SUCCESS in 10m 54s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87fea4dbdcf743d2aea9e4f12e3de270 : SUCCESS in 11m 16s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e0d00985a3f4485a6424650209c2ec5 : SUCCESS in 8m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6da0cc5e7c17449dbdc490413ecd6ba4 : SUCCESS in 1h 10m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/12a8837aba4d43bead200c47bf76e5ef : SUCCESS in 1h 17m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/831b984dcdd04bea882492aa205102ec : SUCCESS in 1h 16m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/098afe3b81b44208a2c956582d25ff3c : SUCCESS in 1h 06m 27s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a94dee363d0492b8aff0a4a6ec06f91 : SUCCESS in 1h 49m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44d6ffce15d149f6b7b5c2db0d9d553a : SUCCESS in 1h 34m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/95e82c16200b4d33b9fde2c85c024c93 : FAILURE in 42m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff42cb9e51464ee792639b3fbb531482 : SUCCESS in 1h 00m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef337afb82094845aa241b962dae7456 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a2cbc561302dcdc93e7ffd220f792450f534149b","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-03-16 14:15:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d7e4dd509b93812310cd8b5ae541fd0b338d189","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-03-16 15:11:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2d3b3e2d8f93f5349a9d5e52df7b3bf0114c8a2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 15:16:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9aafe890787fcbc70e10eb4d776e4c57883a82cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 16:55:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0de7e63772f64cb7bd88ef1f14699640 : SUCCESS in 10m 35s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/809c14acc65f4a99a28e0c9ab2540fc4 : SUCCESS in 10m 24s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef302ee8a3494cd38cc9273d24f22b68 : SUCCESS in 10m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6f987d812117461a9aea7d4d0f142785 : SUCCESS in 1h 17m 20s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3fbff2178415484b876e46e1aa5787fb : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b1291ce37ab4c22a163a485bcd83610 : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f14f8dfc6c67471782eaf822055a528d : SUCCESS in 1h 05m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/65056fcdb782424b8d2f834c5197f0b0 : FAILURE in 37m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/846b598802d6440d9f248eb5f6308486 : SUCCESS in 56m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a82974e69014fa1a67ffbb528be966d : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2aad819a7df6e53a2daa26591dea87f408292981","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 16:55:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7bf9811bb13ed68592c1c7b4b86a9ef5474d6daf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 16:58:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a89e027694a400c9b48a7bf6393d24f : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1132fcc47761326be0c7a42572f0496b1497a79","revisions":{"f1132fcc47761326be0c7a42572f0496b1497a79":{"kind":"REWORK","_number":1,"created":"2021-03-16 08:57:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/45/780745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/45/780745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/780745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/780745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/780745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/45/780745/1"}}},"commit":{"parents":[{"commit":"1ba9f5ea17c4c316e6c5ba2c812edfa45c4ddea7","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/1ba9f5ea17c4c316e6c5ba2c812edfa45c4ddea7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-16 08:57:25.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-16 08:57:25.000000000","tz":0},"subject":"Remove references to unsupported operating systems","message":"Remove references to unsupported operating systems\n\nAll references to Gentoo, SUSE, Debian stretch and Centos-7  are removed.\nConditional tasks, ternary operators and variables are simplified where possible\nOS specific variables files are generalised where possible\n\nChange-Id: I55037805e59cb6d727593f467e98568a4c030e74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/f1132fcc47761326be0c7a42572f0496b1497a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/f1132fcc47761326be0c7a42572f0496b1497a79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
