)]}'
{"id":"openstack%2Fansible-role-systemd_mount~869939","triplet_id":"openstack%2Fansible-role-systemd_mount~stable%2Fxena~I5d2d9c074aa0390a6f8b7c8c0ba1c839fe26f6c5","project":"openstack/ansible-role-systemd_mount","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-01-14 00:20:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5d2d9c074aa0390a6f8b7c8c0ba1c839fe26f6c5","subject":"Fix mount\u0027s systemd unit dependency logic","status":"MERGED","created":"2023-01-12 14:29:43.000000000","updated":"2023-01-14 00:21:37.000000000","submitted":"2023-01-14 00:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869939","cherry_pick_of_change":868511,"cherry_pick_of_patch_set":3,"meta_rev_id":"0cc6be592c9dd532b9a926bf6aa463421a38d14f","_number":869939,"virtual_id_number":869939,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-14 00:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-01-12 14:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-13 22:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-13 22:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-12 14:29:43.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"CC"},{"updated":"2023-01-12 16:36:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-13 22:43:49.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"905c3bd8188d397c757a43ea312ed3668821ca8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-12 14:29:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5ac8832e098b06050ad9babb569ab0311b11ab47","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-12 14:43:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"730c31e7fe8b8d6c0474718ddc06c300a51ba349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 16:36:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e3ced9d51d749abb8ac54e692c7e35a\n\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/5d0fab7af9e148028fe40dfe2f69aae5 : SUCCESS in 7m 37s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1701d5faed06470e9a5b8376670a8658 : SUCCESS in 7m 47s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89ae714eb2cc44e5b06df77e4907d441 : SUCCESS in 6m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fbe156277e424640a8e0eda447568b39 : SUCCESS in 1h 31m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5a484aeaba4647e8b3bc6fff9b780962 : SUCCESS in 1h 13m 48s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ab9ff9dce3b416b8b9cae6cb8e83150 : SUCCESS in 1h 33m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0710178f8dce43ce84d6b8ff5235503b : SUCCESS in 1h 29m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/696083198c344614b64a56f73eb16b55 : SUCCESS in 1h 24m 18s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/af5294c717a546469537ed602b50cc2d : SUCCESS in 2h 03m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/811f8d61b2894b358ee1f58f9b9f248a : SUCCESS in 1h 54m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/e6798b04bd3d46a1adc354f1e1a1398c : FAILURE in 32m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c92c7f5edf14e70b43132675b7b2fb5 : SUCCESS in 1h 04m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dbc7d2fcaa248d19a93ba2ae7d278a8 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":1},{"id":"689fd53e80b78dfa2fc82c90631436d4b9ebc312","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-01-13 22:43:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80868bc261641ca541b22c13ab126ccb463b8120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 22:46:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e527e5390dfd21e29ccef486b776fcefff9923d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-14 00:20:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29e1a5b914f3483f945a2fc6dbd0b2d4\n\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8dd99eaa28d34a78a1e6c18fedd82491 : SUCCESS in 6m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8b60f5b9efec45518f15ce04960e7d69 : SUCCESS in 1h 24m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a99e482fbc144e6fac6e816e7353faa2 : SUCCESS in 54m 53s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/43469af06d9d4592b4237c6af4504ed3 : SUCCESS in 1h 20m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0eb486c27f64b9b837c7a95eec52c55 : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21a7f885bdd240ecaf70e6926b36a259 : SUCCESS in 1h 10m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aec53faacb2a42a59923879bac0f41e3 : SUCCESS in 52m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a6dc381b064de18c3c7a4995edaada : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bb9960c236856317fb7336e3e474ead305afd6fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-14 00:20:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0cc6be592c9dd532b9a926bf6aa463421a38d14f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-14 00:21:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5afe9eb7e8f4e9ebaf8f1b27caf3155\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6453964b551440159751e424a3a27089 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9850f00be4d8ce9bdd8f280e4f433fb9a9617667","revisions":{"9850f00be4d8ce9bdd8f280e4f433fb9a9617667":{"kind":"REWORK","_number":1,"created":"2023-01-12 14:29:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/39/869939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_mount","ref":"refs/changes/39/869939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/39/869939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/39/869939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/39/869939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_mount refs/changes/39/869939/1"}}},"commit":{"parents":[{"commit":"3d346f4c5d8918ad11f7d75ec87050fbd8b9974b","subject":"Update tox.ini to work with 4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/3d346f4c5d8918ad11f7d75ec87050fbd8b9974b"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2022-12-24 00:09:29.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-01-12 14:29:43.000000000","tz":0},"subject":"Fix mount\u0027s systemd unit dependency logic","message":"Fix mount\u0027s systemd unit dependency logic\n\nImprove logic for making systemd mount dependent on\nnetwork-online.target when network filesystem is being used.\n\nThis change improves feature implemented in commit [1].\n\nCurrently when device is specified by UUID or LABEL, dependency for\nnetwork-online.target is added even it\u0027s not a network filesystem.\nWe should be careful with this as it may create circular dependency.\nIn my case, circular dependency looked like this:\n\nnetwork.target -\u003e\n  network-pre.target -\u003e\n    netfilter-persistent.service -\u003e\n      local-fs.target -\u003e\n        var-lib-nova.mount -\u003e\n          network-online.target -\u003e\n            network.target\n\nAdditionally, this commit was originally supposed to add dependency also\nfor CIFS mounts(these ones starting from \u0027//\u0027) but it never worked in my\nopinion. It was also fixed by this change.\n\n[1] https://opendev.org/openstack/ansible-role-systemd_mount/commit/6a9dcb688628a1a3900f25f72effce4f62f6900e\n\nChange-Id: I5d2d9c074aa0390a6f8b7c8c0ba1c839fe26f6c5\n(cherry picked from commit 480bb0c871848f2c65f6c7312eab77c33b88ee7a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/9850f00be4d8ce9bdd8f280e4f433fb9a9617667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_mount/commit/9850f00be4d8ce9bdd8f280e4f433fb9a9617667"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
