)]}'
{"id":"openstack%2Fansible-role-systemd_networkd~658328","triplet_id":"openstack%2Fansible-role-systemd_networkd~master~Ic0196a84e18c6d7fe2e4df40d9b15c3c76aae436","project":"openstack/ansible-role-systemd_networkd","branch":"master","hashtags":[],"change_id":"Ic0196a84e18c6d7fe2e4df40d9b15c3c76aae436","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-05-10 10:48:50.000000000","updated":"2019-08-15 02:17:33.000000000","submitted":"2019-08-15 02:17:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658328-1565835453286-28760403","meta_rev_id":"d00720229168465b90ebfa91ed6df6197cb8f020","_number":658328,"virtual_id_number":658328,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-15 02:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-08-15 00:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-08-15 00:34:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 00:34:14.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-15 02:17:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c2590241c690813e8d4b2ef37e3ffc6a927ce83","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-05-10 10:48:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25badc4211472aff4e0944ea6a4c5c602e2823e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 11:02:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/28/658328/1/check/openstack-ansible-linters/5e32a31/ : SUCCESS in 4m 29s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/28/658328/1/check/openstack-ansible-functional-debian-stable/859f6e3/ : SUCCESS in 6m 55s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/28/658328/1/check/openstack-ansible-functional-centos-7/343b086/ : SUCCESS in 8m 27s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/28/658328/1/check/openstack-ansible-functional-ubuntu-bionic/3add53b/ : SUCCESS in 4m 47s\n- openstack-tox-docs http://logs.openstack.org/28/658328/1/check/openstack-tox-docs/978c5fe/html/ : SUCCESS in 2m 44s\n- build-openstack-releasenotes http://logs.openstack.org/28/658328/1/check/build-openstack-releasenotes/42829a3/html/ : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4a2439990dd36b59c0efcac5cdae0e18ec0ae94b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-15 00:34:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"df6b1d7c50290101dd20de9fe4b60fd0cd2ab66b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 00:34:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ca3dc448353de6f1c2d699f5a6ee7150da17928","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 02:17:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d00720229168465b90ebfa91ed6df6197cb8f020","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 02:17:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c1260b0341024478922c2d7f94bfaf3d : SUCCESS in 4m 22s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/3d5a72555896491088f3d6124a2239ca : SUCCESS in 6m 16s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d44b93c2b38c457184a71ba6c0ddbdb7 : SUCCESS in 9m 27s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ef5d56fff7f4fa1800e388781b577bc : SUCCESS in 5m 59s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b17b60300a2e43fe9702fe187a01d180 : SUCCESS in 1h 11m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/92d9a6eb88cb41e5a9f6cd7475b44bbf : SUCCESS in 1h 28m 00s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/76a79108a6cf4adea5ab04be94127d3e : SUCCESS in 1h 12m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d32f46c6c18427aa01f3d12a01fe034 : SUCCESS in 58m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ac1b47e9e7604ea2baa2565958da9851 : SUCCESS in 1h 03m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4cff9c0c0b74ae4862ed402e82eaaa3 : SUCCESS in 44m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0db3d9bfc043b0865870215f266402 : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6a6e55fbc554f2fbc413ec005a54397 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc954bcb01b5c055bd37a1809b54b8e60cfa5a68","revisions":{"fc954bcb01b5c055bd37a1809b54b8e60cfa5a68":{"kind":"REWORK","_number":1,"created":"2019-05-10 10:48:50.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/28/658328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/28/658328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/28/658328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/28/658328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/28/658328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/28/658328/1"}}},"commit":{"parents":[{"commit":"58a3668db03682bdae608ecea145c5f786e69635","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/58a3668db03682bdae608ecea145c5f786e69635"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-10 10:48:28.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2019-05-10 10:48:48.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: Ic0196a84e18c6d7fe2e4df40d9b15c3c76aae436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/fc954bcb01b5c055bd37a1809b54b8e60cfa5a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/fc954bcb01b5c055bd37a1809b54b8e60cfa5a68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
