)]}'
{"id":"openstack%2Fansible-role-systemd_networkd~660409","triplet_id":"openstack%2Fansible-role-systemd_networkd~master~I6112dcb885f598f218b8a7ba385840a7db343aa3","project":"openstack/ansible-role-systemd_networkd","branch":"master","topic":"use-integrated-repo-for-all-tests","hashtags":[],"change_id":"I6112dcb885f598f218b8a7ba385840a7db343aa3","subject":"Add integrated coverage","status":"MERGED","created":"2019-05-21 14:46:29.000000000","updated":"2019-05-21 22:24:01.000000000","submitted":"2019-05-21 22:24:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660409-1558477441263-a663f06a","meta_rev_id":"5a6c43a39b36ff099df2cb9f2b75620077e1f704","_number":660409,"virtual_id_number":660409,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-21 22:24:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-05-21 15:58:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-21 20:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-21 20:43:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-21 15:58:07.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-21 20:43:16.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-21 22:24:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88d277f57832c825939c29e264edc9b932de36c4","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 14:46:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7f7273238098c988daca4e510edb3fb2c2a8ca3","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-21 14:46:38.000000000","message":"Topic set to use-integrated-repo-for-all-tests","accounts_in_message":[],"_revision_number":1},{"id":"de4a9638b151bd75648fef83330c0cf5c739bb12","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-21 15:58:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5c15a7629df61e141b65c670290e7c1df7a2cdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 16:22:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/660409/1/check/openstack-ansible-linters/e6647e9/ : SUCCESS in 5m 13s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/09/660409/1/check/openstack-ansible-functional-debian-stable/d3c22eb/ : SUCCESS in 6m 15s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/09/660409/1/check/openstack-ansible-functional-centos-7/de30b5e/ : SUCCESS in 8m 25s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/09/660409/1/check/openstack-ansible-functional-ubuntu-bionic/c479f0b/ : SUCCESS in 5m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/09/660409/1/check/openstack-ansible-deploy-aio_metal-debian-stable/8ac06c5/ : SUCCESS in 1h 16m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/09/660409/1/check/openstack-ansible-deploy-aio_metal-centos-7/376dc48/ : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/09/660409/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/e114420/ : SUCCESS in 1h 10m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/09/660409/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/93c4a72/ : SUCCESS in 55m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/09/660409/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5ce9a2a/ : SUCCESS in 54m 28s\n- openstack-tox-docs http://logs.openstack.org/09/660409/1/check/openstack-tox-docs/0432827/html/ : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9914aa5486917ae789fed57c2815df5ef1b5b798","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-21 20:43:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"129a5bbdc027cc6a9b5db7d90efe923d19ebdb61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 20:43:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"871933fe7a303b436c2fc9f59cc7d9d5314bae60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 22:24:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/09/660409/1/gate/openstack-ansible-linters/ce3ad9c/ : SUCCESS in 4m 55s\n- openstack-ansible-functional-debian-stable http://logs.openstack.org/09/660409/1/gate/openstack-ansible-functional-debian-stable/b608e0d/ : SUCCESS in 5m 54s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/09/660409/1/gate/openstack-ansible-functional-centos-7/57f05f0/ : SUCCESS in 7m 23s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/09/660409/1/gate/openstack-ansible-functional-ubuntu-bionic/10d9ede/ : SUCCESS in 6m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/09/660409/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/c6fcecf/ : SUCCESS in 55m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/09/660409/1/gate/openstack-ansible-deploy-aio_metal-centos-7/17abfbd/ : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/09/660409/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/d2be287/ : SUCCESS in 1h 12m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/09/660409/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/32063f5/ : SUCCESS in 57m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/09/660409/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/53c1f8e/ : SUCCESS in 58m 13s\n- openstack-tox-docs http://logs.openstack.org/09/660409/1/gate/openstack-tox-docs/33e1b0b/html/ : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5a6c43a39b36ff099df2cb9f2b75620077e1f704","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 22:24:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08a071fa9cf40bee8a22dfb9dc3e2112ee26efa0","revisions":{"08a071fa9cf40bee8a22dfb9dc3e2112ee26efa0":{"kind":"REWORK","_number":1,"created":"2019-05-21 14:46:29.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/09/660409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/09/660409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/09/660409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/09/660409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/09/660409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/09/660409/1"}}},"commit":{"parents":[{"commit":"30887cf6fa7e95d13b1e79a8f5d75797deebb303","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/30887cf6fa7e95d13b1e79a8f5d75797deebb303"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 14:44:22.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-21 14:46:03.000000000","tz":120},"subject":"Add integrated coverage","message":"Add integrated coverage\n\nCurrently this role doesn\u0027t test the integrated jobs.\nAdding the main integrated jobs make sure this project doesn\u0027t\nbreak the integrated.\n\nThis patch also fixes the meta for the role for openSUSE 15.\n\nChange-Id: I6112dcb885f598f218b8a7ba385840a7db343aa3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/08a071fa9cf40bee8a22dfb9dc3e2112ee26efa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/08a071fa9cf40bee8a22dfb9dc3e2112ee26efa0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
