)]}'
{"id":"openstack%2Fansible-role-systemd_networkd~690171","triplet_id":"openstack%2Fansible-role-systemd_networkd~stable%2Ftrain~Ic2efeffb2e4c918f212ab2777f3d950de16cb178","project":"openstack/ansible-role-systemd_networkd","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"Ic2efeffb2e4c918f212ab2777f3d950de16cb178","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","status":"MERGED","created":"2019-10-22 18:52:20.000000000","updated":"2019-10-24 12:32:34.000000000","submitted":"2019-10-24 12:31:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690171-1571920280828-fe29816b","meta_rev_id":"da7ee5f4d306bfed110dadc299d95c5f2b8c7aa9","_number":690171,"virtual_id_number":690171,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 12:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-10-24 12:15:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-10-24 12:15:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 12:15:32.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-24 12:31:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"942d9da85a82f8adfc234ddbe0e427416c27a150","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-10-22 18:52:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53d760879ebb59acf2f67d1e5b20040f6bc2d313","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:24:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d1b198bd6c574790a4822f3561d5eb0f : SUCCESS in 5m 22s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/0caba81b6fb647a0af1e6e733ce1a4a5 : SUCCESS in 9m 43s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/0d36d60530754a0499a55dbc66ed2d19 : SUCCESS in 10m 58s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddfee9b019024f0da665eb553c50097f : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d30a0a781d3249688e4b760bf149677e : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b572c894a872459b98d9efc0f5eff091 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"7ab2f2c9248d51a9d11c3b006b50158c484e9541","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-24 12:15:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47162d6336686cd729f1b0a3b4c1bbc083e59ccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 12:15:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b0b92becdee7129db6614f6e608315254f723167","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 12:31:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5bebfc87693941dda7e4a94d93e0fadf : SUCCESS in 7m 12s\n- openstack-ansible-functional-debian-stable https://zuul.opendev.org/t/openstack/build/12c5537cb99849f0a25c6da4aa63e5f5 : SUCCESS in 6m 01s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d23d5acf5f074339a9d9f0968227ed73 : SUCCESS in 12m 48s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fc84a054af34f45a85200d6ec6dbc11 : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/601e5d7747a741bba91b32015821a79d : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd632f05828345f297a961f9427222d4 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":1},{"id":"cc56a79e8ca7cbb61ae8e2e91411a643b4eda108","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 12:31:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cd70530afe75e75b8d64d99ff0e14d90c694bf41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 12:32:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7d29b84c6fe48ddb09b930b8cb3ee25 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd0b53415f5744cbb74701a544a49f5f : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43da0e9d58e65043358b260949c6b18baac63288","revisions":{"43da0e9d58e65043358b260949c6b18baac63288":{"kind":"REWORK","_number":1,"created":"2019-10-22 18:52:20.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/71/690171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/71/690171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/71/690171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/71/690171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/71/690171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/71/690171/1"}}},"commit":{"parents":[{"commit":"23ba763b39677a84e718e2b1b491b118b2625825","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/23ba763b39677a84e718e2b1b491b118b2625825"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-22 18:52:17.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-22 18:52:17.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/train branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ic2efeffb2e4c918f212ab2777f3d950de16cb178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/43da0e9d58e65043358b260949c6b18baac63288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/43da0e9d58e65043358b260949c6b18baac63288"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
