)]}'
{"id":"openstack%2Fansible-role-systemd_networkd~727340","triplet_id":"openstack%2Fansible-role-systemd_networkd~master~I88cfdac1d97fb320a99f0b07e101e648ba0b7b4c","project":"openstack/ansible-role-systemd_networkd","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I88cfdac1d97fb320a99f0b07e101e648ba0b7b4c","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-05-12 17:54:13.000000000","updated":"2020-05-12 23:15:29.000000000","submitted":"2020-05-12 23:13:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727340-1589325214198-eaedcf2a","meta_rev_id":"4ec2cce5b1f3d9e6635173e4777f6f16f3030342","_number":727340,"virtual_id_number":727340,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 23:13:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-12 21:21:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-12 21:21:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-12 21:21:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-12 23:13:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a85ad8562d22aab86631b0e3f932280d05fc9d37","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-12 17:54:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4571fbcab80df43690c7d4462d7fd505a2ad07c6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-12 18:06:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8db5a4355167aec739eb9326baaeb98a88fe215","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 21:21:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e929723ae7c9481cb0f6f47f9ce28079 : SUCCESS in 8m 19s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/0578eebc4da54a1698dd681744418f2e : SUCCESS in 10m 03s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/03be5db7d8d04de9b77801538e3cffc6 : SUCCESS in 10m 51s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/97084af58bf64aecb9b0450c51b9a715 : FAILURE in 8m 41s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34ea767f2d6947d79efbe6d038f48ec6 : SUCCESS in 8m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/74075a35303447468046f10c3f1002d1 : SUCCESS in 1h 09m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8c539acc4ab641ba9d78dfa2809e6cb0 : SUCCESS in 1h 30m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/382c969040ec4aa3b6fef73018c28297 : SUCCESS in 1h 29m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d084614a4f984d25bdc7e23458757961 : SUCCESS in 1h 14m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c0785658bcea4068b7e1100edc53287f : SUCCESS in 2h 21m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5a76dfb1cbf4a4abfe3418c889c6cdb : SUCCESS in 2h 03m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e840120435c44e16a57689e77d741c89 : SUCCESS in 1h 12m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/100301d877784e9482aa6beb1eeb6a54 : FAILURE in 36m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/541918bed3bc486bb60eed787b0cafa2 : SUCCESS in 1h 02m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/48b1666c8f8f4b528402d9cc342a1eb9 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/737ae6ed939f40aa97d4ba9049b5ae8e : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d272a0b59be440c5a747de8022087db8 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f835f9dd9ec37a81554258046357b468f1e5db36","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-12 21:21:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e1d6b4ff6903313831e534a24ccac4a9d8ecdfcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 21:21:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c0ef5fc5ffa8d480a847817930770cd0f07834ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 23:13:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/210db9ab459646d1a78a7a9e91ae5a56 : SUCCESS in 8m 54s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/9d6612b31308444b90f641940bc5c4bf : SUCCESS in 8m 23s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/67b27cfb4b644ae88cf3b18e91bdb811 : SUCCESS in 12m 25s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/295f53f10dc74279af7f97d82e9b3822 : SUCCESS in 9m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e6b60b7c6f954e56916cff5fcdfcbf2f : SUCCESS in 1h 06m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a5d7153803c348418b48ee318ecc572d : SUCCESS in 1h 43m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/98ac7f22c1684d009d28c6f345b53c17 : SUCCESS in 1h 31m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1f0cb689a31474cb241580071f0407d : SUCCESS in 1h 04m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/404f84fa6925460d806bc067407039c4 : SUCCESS in 1h 12m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce2f4ce0e4834a62a4b184ee26a54a9c : SUCCESS in 56m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/541e406c202c4e0799619c6e5dada4fd : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/571b7d9632b24c0ea634bee13a4df993 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8228c69d39c24fde9e8e9c5edeea3da2 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"aeed45a2ab93f8af8ad7f9865e3e65adecb71c4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 23:13:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6badc19b8ab927881470e28d58a543f9206cf675","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 23:15:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/981d6dfda4c94ab4a10c4159fbe8de63 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54053c10dc6941c9be4158d27dc37450 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2c77e78f22c4df3e6ef41b8884f248749ca55650","revisions":{"57666f10196095cb4326cfb402c03909d0a01f38":{"kind":"REWORK","_number":1,"created":"2020-05-12 17:54:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/727340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/40/727340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/40/727340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/40/727340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/40/727340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/40/727340/1"}}},"commit":{"parents":[{"commit":"b2533221f4470cc5b790c08df0fe750a3004e6fc","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/b2533221f4470cc5b790c08df0fe750a3004e6fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 17:54:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 17:54:11.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I88cfdac1d97fb320a99f0b07e101e648ba0b7b4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/57666f10196095cb4326cfb402c03909d0a01f38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/57666f10196095cb4326cfb402c03909d0a01f38"}]},"branch":"refs/heads/master"},"2c77e78f22c4df3e6ef41b8884f248749ca55650":{"kind":"REWORK","_number":2,"created":"2020-05-12 18:06:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/727340/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/40/727340/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/40/727340/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/40/727340/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/40/727340/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/40/727340/2"}}},"commit":{"parents":[{"commit":"b2533221f4470cc5b790c08df0fe750a3004e6fc","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/b2533221f4470cc5b790c08df0fe750a3004e6fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 18:06:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-12 18:06:08.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I88cfdac1d97fb320a99f0b07e101e648ba0b7b4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/2c77e78f22c4df3e6ef41b8884f248749ca55650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/2c77e78f22c4df3e6ef41b8884f248749ca55650"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
