)]}'
{"id":"openstack%2Fansible-role-systemd_networkd~736201","triplet_id":"openstack%2Fansible-role-systemd_networkd~stable%2Fussuri~I5c5eba38917254187a5b006fd810977ee48ec023","project":"openstack/ansible-role-systemd_networkd","branch":"stable/ussuri","topic":"interpreter-stable/ussuri","hashtags":[],"change_id":"I5c5eba38917254187a5b006fd810977ee48ec023","subject":"remove ansible_python_interpreter","status":"MERGED","created":"2020-06-17 13:09:57.000000000","updated":"2020-06-18 08:25:03.000000000","submitted":"2020-06-18 08:22:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736201-1592468552356-22558d1a","meta_rev_id":"c75cad61d20b089cdeb578f54186752699485088","_number":736201,"virtual_id_number":736201,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 08:22:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-06-17 20:20:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-18 06:23:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-18 06:23:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 20:20:30.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-18 08:22:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a92f35ce2aa181a9b511e099b62daa93b31490eb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 13:09:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"06b97119cecf6a18d12bd2000b28e6a1dbfd79aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 16:02:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/98aebcddd38c4fd78ad864be0e651509 : SUCCESS in 10m 42s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/b285d79b3f6045e38eca33ada1c8dc50 : SUCCESS in 11m 57s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/9ad4c1986d804767939af7ba47bbb660 : SUCCESS in 14m 26s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/cdf4aad7782d41b1aef76b631b000a4f : FAILURE in 13m 39s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29fae24c8d4d46a3b8440ac3d6b0f8e1 : SUCCESS in 13m 03s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c172534a4914ff5b4f893c88c4e2fd0 : FAILURE in 13m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a5febfdead344b5abaac3af25ddbcafb : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2d117098b96348d2b502186f83436a57 : SUCCESS in 1h 35m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12afe46a7ee24aa4837ee607a4d027c5 : SUCCESS in 1h 32m 26s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e0e54f686bb7490aa7bf60746af1ac3c : SUCCESS in 2h 40m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17d424f3b43d40068c739486ecb62e5f : SUCCESS in 1h 57m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/447fc7ab33554146901971a3b607cb19 : SUCCESS in 1h 44m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d895d6219714fefa0144e6add66b8f5 : SUCCESS in 59m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbb14ee65394dc99741169f3db46e45 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9e39084fc8a6461c9a89cfcd77818da30f3b9076","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 16:05:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6d32e656a126c9aa4105741309f50cc967a0bced","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-17 19:58:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7c6e06deaefd31e8b8fa5d36a7d2030a5080f741","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:59:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81d6e071f28707f11af72288eb0d3b62d722d260","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 20:10:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6a0fd6c91638631ef44c9deadb97b3b45d0e9962","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-17 20:10:31.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"64bae109d76b8fa34148fb2abf9056ab71da9e40","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-17 20:20:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ca23a27863174a5931b683e615684515cfeac9ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 22:52:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/046c567802094260b16d7b8de3c086d0 : SUCCESS in 10m 22s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/e11f5d8f6ae247a78df7f3f646fdaead : SUCCESS in 13m 54s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/1658fd7404b9497bb5769059e4de9e46 : SUCCESS in 19m 14s\n- openstack-ansible-functional-centos-8 https://zuul.opendev.org/t/openstack/build/d4bec3aba4ef47068c366a3f06210bc6 : FAILURE in 11m 25s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e46cddc2b684de292d024503603ca29 : SUCCESS in 9m 00s\n- openstack-ansible-functional-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9421b97165004ff389e3e869d677a408 : FAILURE in 14m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/68e07c3511384a00b51d893d454712ab : SUCCESS in 1h 13m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/518b5bd6bfe343e49488de345126f563 : SUCCESS in 1h 56m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fba1ad9aa9604572aa095c6229a7aaeb : SUCCESS in 1h 29m 49s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3639dbf794f649e8b552810a190f2ff9 : SUCCESS in 2h 25m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/838125f9e9494707b4f9f7718cbcd08b : SUCCESS in 2h 02m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ecfb3f6393854ffdbb1d60a03103f60b : SUCCESS in 1h 17m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b2501e2d3af4505a2dba0eff1e4faec : SUCCESS in 46m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d97ac4c9da4400f8eb62152c1062701 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":3},{"id":"c1bb294934de1799534cbf76e2bb251152259a88","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-18 06:23:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ea706e19a63ffbc8d90c12de11a1a53e9dc80900","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 06:24:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"35e38ec48f90cd39e6162f2c934d26362bbb07a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:22:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/61ee02c9d9b643ca88411f5fab0398c3 : SUCCESS in 13m 21s\n- openstack-ansible-functional-debian-buster https://zuul.opendev.org/t/openstack/build/70dea0ee18784928841f7fbe8656706a : SUCCESS in 13m 45s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/4dfc371f153f42db8a9e560ea4f07fc1 : SUCCESS in 12m 03s\n- openstack-ansible-functional-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6837e05dc2854131bdf49efa26ea14fb : SUCCESS in 10m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/da4ce760a736437a8c02519a663c8863 : SUCCESS in 1h 08m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3c80276c7c18465fa203f056c65f749f : SUCCESS in 1h 45m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/142e2c46a741445cb25f797a8b7c2a63 : SUCCESS in 56m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e3e111e0478146fe91e089391cb09430 : SUCCESS in 1h 10m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9db696a7022e495f88ebe39469d85232 : SUCCESS in 1h 01m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace066171b144a3e8c3793f241f1610d : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b13108c43fc5812ab6263a1f25707a4666828f85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:22:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"48ebadac13e88323a00f81600425971057d65c37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:25:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76958144f81946d782a3251d9333522c : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bc5a64964f2a9fd8ed6f31ab43cee33083018326","revisions":{"1c54e9f384b339ec479b263fd8b1d60314b32c1b":{"kind":"REWORK","_number":1,"created":"2020-06-17 13:09:57.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/736201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/01/736201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/1"}}},"commit":{"parents":[{"commit":"138e8a091661f645c9e4ed9324baee564ea6dc39","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/138e8a091661f645c9e4ed9324baee564ea6dc39"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 17:51:58.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-17 13:09:57.000000000","tz":0},"subject":"remove ansible_python_interpreter","message":"remove ansible_python_interpreter\n\nansible_python_interpreter is now set to auto in the tests repo.\nit doesn\u0027t need to be overwritten in the test anymore.\n\nDepends-On: https://review.opendev.org/734676/\nChange-Id: I5c5eba38917254187a5b006fd810977ee48ec023\n(cherry picked from commit 15a0be8cbb725c42b201dcf54f95b309b9b13305)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/1c54e9f384b339ec479b263fd8b1d60314b32c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/1c54e9f384b339ec479b263fd8b1d60314b32c1b"}]},"branch":"refs/heads/stable/ussuri"},"b4e8267576bc0b425912c5ae502b969179db81b0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-17 20:10:23.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/736201/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/01/736201/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/2"}}},"commit":{"parents":[{"commit":"138e8a091661f645c9e4ed9324baee564ea6dc39","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/138e8a091661f645c9e4ed9324baee564ea6dc39"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 17:51:58.000000000","tz":-240},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-17 20:10:21.000000000","tz":0},"subject":"remove ansible_python_interpreter","message":"remove ansible_python_interpreter\n\nansible_python_interpreter is now set to auto in the tests repo.\nit doesn\u0027t need to be overwritten in the test anymore.\n\nDepends-On: https://review.opendev.org/736198\nChange-Id: I5c5eba38917254187a5b006fd810977ee48ec023\n(cherry picked from commit 15a0be8cbb725c42b201dcf54f95b309b9b13305)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/b4e8267576bc0b425912c5ae502b969179db81b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/b4e8267576bc0b425912c5ae502b969179db81b0"}]},"branch":"refs/heads/stable/ussuri"},"bc5a64964f2a9fd8ed6f31ab43cee33083018326":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-06-17 20:10:31.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/01/736201/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/01/736201/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/01/736201/3"}}},"commit":{"parents":[{"commit":"138e8a091661f645c9e4ed9324baee564ea6dc39","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/138e8a091661f645c9e4ed9324baee564ea6dc39"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2020-06-11 17:51:58.000000000","tz":-240},"committer":{"name":"Marc Gariépy (mgariepy)","email":"gariepy.marc@gmail.com","date":"2020-06-17 20:10:26.000000000","tz":0},"subject":"remove ansible_python_interpreter","message":"remove ansible_python_interpreter\n\nansible_python_interpreter is now set to auto in the tests repo.\nit doesn\u0027t need to be overwritten in the test anymore.\n\nChange-Id: I5c5eba38917254187a5b006fd810977ee48ec023\n(cherry picked from commit 15a0be8cbb725c42b201dcf54f95b309b9b13305)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/bc5a64964f2a9fd8ed6f31ab43cee33083018326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/bc5a64964f2a9fd8ed6f31ab43cee33083018326"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
