)]}'
{"id":"openstack%2Fansible-role-systemd_networkd~869736","triplet_id":"openstack%2Fansible-role-systemd_networkd~master~Iee71e9839d4c8d318c2c36152d966dd5a1b88111","project":"openstack/ansible-role-systemd_networkd","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2023-01-11 19:55:38.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-01-11 10:59:25.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Iee71e9839d4c8d318c2c36152d966dd5a1b88111","subject":"Handle omitted variables which appear as empty strings","status":"MERGED","created":"2023-01-10 15:07:56.000000000","updated":"2023-01-11 19:56:32.000000000","submitted":"2023-01-11 19:55:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"869736","meta_rev_id":"de12bb0589f5eb0c1be0af1d79b8df76f1f3fc4a","_number":869736,"virtual_id_number":869736,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-11 19:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-01-11 15:53:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-11 15:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-11 15:55:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-01-11 08:09:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-10 15:08:32.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-01-10 15:46:46.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-01-10 17:33:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-11 15:55:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"8d608fb43d15c12908319dffc00938806dfacd0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-01-10 15:07:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5283fb702a93dcd0ff6a032c648a955d6c78bc7d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-10 15:08:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c60e6edeb2e3b32d187b2877bd56d4b877700f91","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-01-10 15:08:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"043c3f40698654c24c58afbcdb003e224c8232c0","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-01-10 15:11:33.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"380be38e1dd9c1fb8234401bf4aa1a98d0a32dd8","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-01-10 15:38:30.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"c04b71a429588f4edf3204d2db63da3dc01a406d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-01-10 15:45:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2759e7aa16c87cc4ee1b079752f4c67479fb1cd2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-01-10 15:46:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aebd66d7e59cb05fbbd83d4484f247b161afdf56","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-10 15:52:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d222937ae7694ac091e48d79b6b3f48bf71d9996","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-01-10 15:55:15.000000000","message":"Patch Set 2: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7ac97aba4f89ffc6aed531b31c31acc4802f57a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 17:33:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/646b68fd17bd402cb37ed5f0ea2c9c27\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5248f188e47447c98354c8b1e581c733 : SUCCESS in 15m 32s\n- openstack-ansible-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/be340502fed74fae863ae494a019772c : SUCCESS in 12m 54s\n- openstack-ansible-functional-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84b315669c4d4334a9fe7243e32d8b48 : SUCCESS in 10m 25s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a1658b2b6ed443b8963300d7a5a09ff9 : FAILURE in 40m 37s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/632c026e173a4b189ad667922aca9062 : SUCCESS in 1h 22m 53s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8e396d9b776643c7825ba7c91c9e31ee : SUCCESS in 1h 27m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/956daa7ed7c143739fa4de7ee13811a2 : SUCCESS in 1h 25m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/734953382af54583bc1dfb880b5de8c5 : SUCCESS in 1h 21m 19s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/6cb86cc664834e7fb20197b816c0137c : SUCCESS in 1h 35m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c2c1f7576a34fd992442b28ba41ddaf : SUCCESS in 1h 44m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/22c68edfb34d4e6e8ab94196ae08f261 : FAILURE in 37m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3d2bb9317284b9f8cd78dd795162379 : SUCCESS in 58m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f0f75264fce4277a5d08a95319e209c : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"e5ed1d125c8f2818d9bf460fea5004ef49cd5a20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-01-11 08:09:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a4e91a4ab89ceb8eaf5684793119a60e5940ba54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 10:12:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2df1a2bc7dd46ecacf61f07e1362ea4\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dace73113fb6433ca01bc8b420fb6bb9 : SUCCESS in 10m 25s\n- openstack-ansible-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8032c38bc4d49ceb3c5c49f04858665 : FAILURE in 9m 05s\n- openstack-ansible-functional-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84774adf4f8d41a987935e0e2b093635 : SUCCESS in 7m 59s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a533753868074221bb661134f38aeb88 : SUCCESS in 1h 16m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e09c534acfd043dd8f0be669b1b1faca : SUCCESS in 1h 34m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d2089726853c4dc5a727c84c84c133f8 : SUCCESS in 1h 19m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/891bf1bf5b9f437fa8143d0e48bf8a72 : SUCCESS in 1h 42m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52315b0fa3c64a77ac3f4d0a5601ad9e : SUCCESS in 1h 22m 01s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/32430f769c2246ea95fd668c4cd04d62 : SUCCESS in 1h 38m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b98537ec7d041278edba4b218bad14f : SUCCESS in 1h 56m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/58a09c02b1144cc9a86acdc13927ecda : FAILURE in 43m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f126b0d18c2443389c2e844738424fb0 : SUCCESS in 56m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d12c93b97f46999fdb01d1a2d9109c : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":3},{"id":"77e603058d4b9f3fbdc2c031d227813205d567f7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-11 10:59:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da23b92862034538f2f3728b8f59c3ebc43deaa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 12:58:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62bf32cd23f745ab91e1c61fd4d9ca83\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d2584968f694624960c8a2bcf84ae1b : SUCCESS in 13m 09s\n- openstack-ansible-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6bd78845c634c36abf6922cea532d61 : SUCCESS in 13m 24s\n- openstack-ansible-functional-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af6f24d669e1495294fee40ea58dea10 : SUCCESS in 9m 23s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c11bc1b15bb745ed9346d3f0ddb50f23 : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/59093eab0dc74171b3ea56e9fa533f6e : SUCCESS in 1h 31m 40s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eb475d44c8954906b939dd69904468f6 : SUCCESS in 1h 36m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d868a14afc5e4d549a1cd6e6d1c0d8e2 : SUCCESS in 1h 41m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/160f92630b7d41f3afa8ba058ebc48e3 : SUCCESS in 1h 34m 23s\n- openstack-ansible-upgrade-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a2feccb7a21464aac1dfb0e44180aa2 : SUCCESS in 1h 55m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdf46e725f6c41698e03b2ab7cabfaf9 : SUCCESS in 1h 54m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd53468d46ec4f41af237bda85d1c0ba : FAILURE in 44m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43f21b8fd1f946309ed2fbb2e474dc29 : SUCCESS in 1h 20m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344584f403b24f5e9963b5909cb5e41a : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"7cd9e485a362f4eb4b041877b39e70d8733faa39","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-01-11 15:53:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a0246dbdbca83bed8eac099539a19203c3b4e492","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-01-11 15:55:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c9169a892864b75b190cff81daca963c2a4936ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 18:34:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8864c0a5bdeaf9bf2f406efd04972fdce022caf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 19:55:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f28a68c56ea7422eaf8d7316fe2b1a0e\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f46c5cef76b4744808c9a07e7dccf1c : SUCCESS in 15m 25s\n- openstack-ansible-functional-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aea6e663486469abddb34f6f9978599 : SUCCESS in 7m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9daf8382b92949fe8f01274061400dd3 : SUCCESS in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef823ee3ad684abd88f5475eea627cb5 : SUCCESS in 1h 16m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e72f8fe903f94ba8afd97d53900e6ecc : SUCCESS in 1h 05m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb1a29375b334e1784c269ed73d17c93 : SUCCESS in 1h 16m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9457521b236c4819853721f29ae13305 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":3},{"id":"c57543c10d7dd4b3587df252e44f254558a0efa7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 19:55:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"de12bb0589f5eb0c1be0af1d79b8df76f1f3fc4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 19:56:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/765c38d05123443cb230b6dd050d6bef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412bee4351a6494b95fd2f9dc671d237 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4a7012e366ba60b99ec022eafbfe23afc9f6247a","revisions":{"bc673e3dee2c2e6c93f6277cf5a7b29c2fdc2a15":{"kind":"REWORK","_number":1,"created":"2023-01-10 15:07:56.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/36/869736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/36/869736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/1"}}},"commit":{"parents":[{"commit":"5acf51aec34005ef756b917c77874e5f14fa7ae3","subject":"Fix static routes to use Destination rather than Source key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/5acf51aec34005ef756b917c77874e5f14fa7ae3"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2023-01-10 15:05:13.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2023-01-10 15:05:13.000000000","tz":0},"subject":"Handle omitted variables which appear as empty strings","message":"Handle omitted variables which appear as empty strings\n\nChanges to external roles cause some variables to be passed in\nas empty strings which result in invalid config and InvalidArgument\nerrors.\n\nThis could be addressed here or in the calling role.\n\nChange-Id: Iee71e9839d4c8d318c2c36152d966dd5a1b88111\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/bc673e3dee2c2e6c93f6277cf5a7b29c2fdc2a15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/bc673e3dee2c2e6c93f6277cf5a7b29c2fdc2a15"}]},"branch":"refs/heads/master"},"8fcbc0a8fcf4ad6725910697198bf541c297fa53":{"kind":"REWORK","_number":2,"created":"2023-01-10 15:45:18.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/36/869736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/36/869736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/2"}}},"commit":{"parents":[{"commit":"5acf51aec34005ef756b917c77874e5f14fa7ae3","subject":"Fix static routes to use Destination rather than Source key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/5acf51aec34005ef756b917c77874e5f14fa7ae3"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2023-01-10 15:05:13.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2023-01-10 15:44:33.000000000","tz":0},"subject":"Handle omitted variables which appear as empty strings","message":"Handle omitted variables which appear as empty strings\n\nChanges to external roles cause some variables to be passed in\nas empty strings which result in invalid config and InvalidArgument\nerrors.\n\nThis could be addressed here or in the calling role.\n\nChange-Id: Iee71e9839d4c8d318c2c36152d966dd5a1b88111\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/8fcbc0a8fcf4ad6725910697198bf541c297fa53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/8fcbc0a8fcf4ad6725910697198bf541c297fa53"}]},"branch":"refs/heads/master"},"4a7012e366ba60b99ec022eafbfe23afc9f6247a":{"kind":"REWORK","_number":3,"created":"2023-01-11 08:09:29.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/36/869736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/36/869736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/36/869736/3"}}},"commit":{"parents":[{"commit":"5acf51aec34005ef756b917c77874e5f14fa7ae3","subject":"Fix static routes to use Destination rather than Source key","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/5acf51aec34005ef756b917c77874e5f14fa7ae3"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2023-01-10 15:05:13.000000000","tz":0},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2023-01-11 08:08:37.000000000","tz":0},"subject":"Handle omitted variables which appear as empty strings","message":"Handle omitted variables which appear as empty strings\n\nChanges to external roles cause some variables to be passed in\nas empty strings which result in invalid config and InvalidArgument\nerrors.\n\nThis could be addressed here or in the calling role.\n\nChange-Id: Iee71e9839d4c8d318c2c36152d966dd5a1b88111\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/4a7012e366ba60b99ec022eafbfe23afc9f6247a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/4a7012e366ba60b99ec022eafbfe23afc9f6247a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[]}
