)]}'
{"id":"openstack%2Fansible-role-systemd_networkd~908813","triplet_id":"openstack%2Fansible-role-systemd_networkd~stable%2F2023.2~If82946b7f5ad3c618606110576c586632f9bbd08","project":"openstack/ansible-role-systemd_networkd","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-02-24 00:04:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If82946b7f5ad3c618606110576c586632f9bbd08","subject":"Use OriginalName instead of Name in systemd.link","status":"MERGED","created":"2024-02-13 14:57:18.000000000","updated":"2024-02-24 00:04:59.000000000","submitted":"2024-02-24 00:04:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908813","cherry_pick_of_change":907527,"cherry_pick_of_patch_set":2,"meta_rev_id":"65ad0ae7f6b909c1fd8dd2615db222e62712e11e","_number":908813,"virtual_id_number":908813,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-24 00:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-02-21 10:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-23 22:24:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-23 22:24:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-13 18:00:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-20 15:31:21.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2024-02-23 22:24:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"685039e2b0b5c75415d100b35c707d0c8fe68985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-13 14:57:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c3c26bcab84abcfd26595982569ee700db8915f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:00:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dd3c824fe824df39f70a1324f6b90f8\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9335498e0794608b610bbb8b0f5ce6f : SUCCESS in 12m 57s\n- openstack-ansible-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d282823054f4030a6ca37eb6bebfe5a : SUCCESS in 9m 46s\n- openstack-ansible-functional-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16cbb5a68cd145ed83c051cff00cb48e : SUCCESS in 7m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/8b6bf60a42304482b443c429c28fba95 : SUCCESS in 1h 38m 03s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/612dc683422945c1abde4a328c725bcb : SUCCESS in 1h 06m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b439ddbc18cf4c009475ef651baa0c18 : FAILURE in 1h 10m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1ab2fef3b2e445cebed603fffcda2ab4 : SUCCESS in 1h 29m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eebb900fb835470da35aaf6aaf21f8ba : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/add7990c4afa4c529f21d12913f55aa3 : SUCCESS in 1h 25m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/957fb3328ada43868d1fdd5341519f8a : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/348f8f18b0d74397b106d20dc0faf557 : SUCCESS in 1h 31m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e9a2dc496868497e939298459f86e7ed : SUCCESS in 2h 07m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8f1b9a811e546ff8ab224a434a406b6 : FAILURE in 1h 19m 21s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2e5d73647d734f53ab76eb22dc0006c7 : SUCCESS in 2h 25m 39s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3695f39ade1942538558ffb87fc470db : SUCCESS in 2h 50m 31s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/176bf81d969441dabcb2f567c149e9a9 : SUCCESS in 1h 06m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38106793ed98425789204b160360be77 : SUCCESS in 1h 02m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d56a2e63edbe484d955bdfb8bf35f814 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fbaf16702aac94fc2d8b889a0e384ac0acbfe761","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-02-20 15:31:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e2b63e22a490ee3d688d37ca1d31e32ef3cee707","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-20 17:48:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eff77fb862ec9b2f8f02bb4b57837d33379f2c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 18:15:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc428842636049aead6ece4592629f1f\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0efc16756016402eb86d2c7d468c178e : SUCCESS in 13m 52s\n- openstack-ansible-functional-centos-9-stream https://zuul.opendev.org/t/openstack/build/6787ef2b2b3f47d2b7e9a1d94514cea5 : SUCCESS in 10m 17s\n- openstack-ansible-functional-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9950d5c2215141cdb27b51d74238a76d : SUCCESS in 7m 48s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/3aed0b8b3724425a81a0c9e4c5f1e79e : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/adc9e3b8acb94586875b7084315fbac6 : SUCCESS in 1h 20m 07s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d138f262a0cb43d0ad5b3d963423996e : SUCCESS in 1h 26m 42s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/51460dfdd14f4b6aabfe495eafd9603a : SUCCESS in 1h 28m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbe65b83e06648be989405060ed54f76 : SUCCESS in 1h 26m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b01e61103a84b3fa79f39b37f3cb937 : SUCCESS in 1h 18m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfec1d67c7314d9a8844037fb63f99c7 : SUCCESS in 1h 19m 19s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d980d302f364e7ab5281c3373aefa45 : SUCCESS in 1h 28m 54s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb4079f1a5d845d886f0c8b0a3ceed86 : SUCCESS in 2h 12m 10s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/111667d0a61c41cdb79a25839da808ef : SUCCESS in 2h 32m 18s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ba6f19815084d4f800aca8cee1b72bd : SUCCESS in 2h 22m 36s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4738e0414334ce8b14cbaa5ca9b8fd6 : SUCCESS in 1h 53m 58s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/31ea6c5bee8a498dba6c93b25ca846d8 : SUCCESS in 1h 07m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1b4d98021364e87baa0439bdab20c19 : SUCCESS in 1h 27m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac16d9e0840a4ec6bf4b591ae47d6a2e : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"26d3bd8274a4487a8251c8fd6e352e363171c559","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-02-21 10:45:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2c9d5d063cff18f68ad02ac64dc06192e4349d48","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-02-23 22:24:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb137d96cac557e44f4a9fc933192607877f3170","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 22:26:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"093994877596af614eddb405a521f5452ee326c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 00:04:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7e5d0987dae42e4843c60005bba76d2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18b6c432a84a4fd385f7eb7ff7ddf119 : SUCCESS in 16m 14s\n- openstack-ansible-functional-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e65bec834e3348278ad089fe9948b8e6 : SUCCESS in 7m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/feb6d3d50caf4b1da659818f099ec004 : SUCCESS in 1h 09m 56s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9210a05d512a4a6b92086242043de430 : SUCCESS in 1h 10m 10s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f65a68c9bf9d43a3bda1960a4473ba19 : SUCCESS in 1h 21m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1af39c5057c34e8ea81300e76b1f3cb1 : SUCCESS in 1h 06m 05s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5555e208124c453b9f16958415dd694e : SUCCESS in 1h 07m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/290b4afa66204633801238cff2829eb0 : SUCCESS in 1h 11m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bf2e07f23904a23a8b21eadc1f38263 : SUCCESS in 1h 34m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9852a2bce2d842b5a152bb070532cf26 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"046d6a5b3871ac987888e2bbb05fc3064c3a76b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 00:04:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"65ad0ae7f6b909c1fd8dd2615db222e62712e11e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 00:04:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c81309d16a6c4d7ca7bcf2084c2e6a29\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eca7d913797403cbf7e5005ffeb4d46 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aafc7ba7085410c80a28b02be5c612da11a4e805","revisions":{"aafc7ba7085410c80a28b02be5c612da11a4e805":{"kind":"REWORK","_number":1,"created":"2024-02-13 14:57:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/13/908813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/13/908813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/13/908813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/13/908813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/13/908813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/13/908813/1"}}},"commit":{"parents":[{"commit":"a05e610cc53239f0c62dc0aae64d00f1d8e75be0","subject":"Fix defenition of multiple static routes for network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/a05e610cc53239f0c62dc0aae64d00f1d8e75be0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-02-02 11:16:14.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2024-02-13 14:57:18.000000000","tz":0},"subject":"Use OriginalName instead of Name in systemd.link","message":"Use OriginalName instead of Name in systemd.link\n\nName option is not a valid one for systemd.link [1] while it is for systemd.network [2]\nthough having same meaning.\n\nThis basically resulted in link files being just ignored by udevadm\nand were not loaded properly after reboot\n\n[1] https://www.freedesktop.org/software/systemd/man/latest/systemd.link.html#OriginalName\u003d\n[2] https://www.freedesktop.org/software/systemd/man/latest/systemd.network.html#Name\u003d\n\nCloses-Bug: #1896589\nChange-Id: If82946b7f5ad3c618606110576c586632f9bbd08\n(cherry picked from commit b9c7d79089478d115e18d29c9ae9de0cbd49d58d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/aafc7ba7085410c80a28b02be5c612da11a4e805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/aafc7ba7085410c80a28b02be5c612da11a4e805"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
