)]}'
{"id":"openstack%2Fansible-role-systemd_networkd~971745","triplet_id":"openstack%2Fansible-role-systemd_networkd~master~Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29","project":"openstack/ansible-role-systemd_networkd","branch":"master","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-07-14 23:09:28.000000000","reason":"Change was submitted"},"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-07-14 09:47:40.000000000","reason":"\u003cGERRIT_ACCOUNT_32666\u003e replied on the change","reason_account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-06-25 08:50:19.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-07-14 18:07:50.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-07-14 23:09:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29","subject":"Remove support for RHEL 7 and SUSE 42","status":"MERGED","created":"2025-12-23 23:28:33.000000000","updated":"2026-07-14 23:10:24.000000000","submitted":"2026-07-14 23:09:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971745","meta_rev_id":"fff3d4cbf6db5b8e209f518985a3a17b76c3f5a1","_number":971745,"virtual_id_number":971745,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-14 23:09:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2026-07-14 09:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":2,"date":"2026-06-25 08:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":1,"date":"2026-07-14 09:47:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-24 01:40:25.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2025-12-24 02:36:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-12 16:58:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-01-12 16:58:20.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-06-11 15:41:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"},{"updated":"2026-06-11 15:41:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"bf4cf8037178e879922a8b55b87b62088eacd388","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 23:28:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4390811ba8d194e49c8a32df0c953c64f2c9bd89","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 23:34:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ab5f9544e58deb65205990ed54b25155391dcb50","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 23:35:09.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3dfcb0821a22442abd24341824f313d9e18bb6d6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 23:35:13.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"bb1c8db7071280ff1fb273f8329a158818c76a08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 23:39:09.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8b574c37c38ee702ae6551dc7e14af3d3f10c4d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-24 00:00:50.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"e119cef69fb934b72864810308931af687e8ef0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-24 00:03:25.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"4af119dfd88f06bd83872098819f4a22cebacce4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-24 00:21:03.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"4fe1a8f12b9791c91e2c039346b3106d31499f54","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-12-24 01:40:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"20085f3b1b6653814b19deddddce28949f835e69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 02:36:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68235a5eacec46feb40743b5d26d45cd\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/40e54b283ef5474ba3c37ece965f33d0 : SUCCESS in 4m 23s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/6ccf9b62ab474690b213d35784664b04 : SUCCESS in 8m 43s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/0fd0070f37214b349ab390557c827aba : SUCCESS in 4m 01s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c6d7cb8573a2493696727529765ad5d3 : SUCCESS in 7m 42s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/357bec8b1a8a40e39f77b5f3c8d43024 : SUCCESS in 6m 43s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ce1cb98966fd4e4fa5d99c1360718907 : SUCCESS in 6m 47s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/13d1b00a8f4f4fb28c569b803015c011 : SUCCESS in 40m 23s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/26c8737fe20447758c69903653dcb0d1 : SUCCESS in 42m 28s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/866a9ca22942417eba35fb2938e4c957 : SUCCESS in 1h 22m 12s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6f29b00c185242a9b0dd3bffad8f5234 : SUCCESS in 1h 08m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4b1563f540cf478d838023fb089d9e22 : SUCCESS in 1h 30m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/432db73f50d24d43871b80874e0ce7bb : SUCCESS in 41m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a24c4f793981480681d1f3c667d145d3 : SUCCESS in 1h 24m 57s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/87464ec09fda4f9db529a11f53a94777 : SUCCESS in 1h 45m 15s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7aab2c88c4a74f4ca09917493569cc0e : SUCCESS in 1h 45m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9cb3238b749c493797796cef7e6cae29 : SUCCESS in 2h 02m 28s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/90123229fe064982aa095a3a7d2c37f0 : SUCCESS in 2h 08m 07s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/35f777e7e1a842f399f3ea65694620d0 : SUCCESS in 2h 06m 30s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/45ffad1dbc5241e5ae70315cd019d289 : FAILURE in 8m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a68ec51ae1b747569dba74af59e13a43 : FAILURE in 55m 15s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ce23cabe8ac4f1482c8ffa7e0e864f1 : SUCCESS in 53m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf7baa10fc14421a07d3506b4466ece : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":7},{"id":"6d8ad0a63d6aab0522e4e041e1cf3e06adaf77d7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-24 08:05:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7},{"id":"01d04ef5bf54f1dc1c50ddf958e7d53186f46c67","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 16:56:03.000000000","message":"Restored","accounts_in_message":[],"_revision_number":7},{"id":"5a9a81e98e46b18d2b079790a2556906a0d6dd3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 16:58:05.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"200c522690af6459ba8427ab79729736b6ecacf7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 16:58:11.000000000","message":"Topic set to osa/bindep","accounts_in_message":[],"_revision_number":8},{"id":"5fea322f1b85f0a745e24512a73f6e8bd4dd4144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 19:34:26.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb5652d0e7654a5fbb019342e29fb7d6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9b4add1d6c943f4bf7253b9620a7ba0 : SUCCESS in 12m 54s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/0efc58de471e492cb7d506990ff487ac : SUCCESS in 7m 47s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/8d55417dab9946dc8775cdd6e943c14f : SUCCESS in 4m 47s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69f71d37e4024e50877b0de9d3908c0a : SUCCESS in 8m 04s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/5f8b6f87d8ae4549b8077def0a4910c4 : SUCCESS in 7m 19s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fdbcbe0e01264d21b09f2fdb4183a401 : FAILURE in 5m 19s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/a0c52f89589044daa9b78f43c9a7f530 : FAILURE in 18m 51s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c8c19afaf154fa58e9afbb3e7668880 : FAILURE in 36m 12s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/48daac8fb1a64d45b56618f3c8d6312e : SUCCESS in 48m 05s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/71d1c83035f848318a6165dbc77e11c7 : FAILURE in 19m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/eb6ec5a0c3c74d67a826219cc41e9723 : FAILURE in 18m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6dfb847d99324f06b98cbafa614a8ce2 : SUCCESS in 47m 51s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36067b6e445146cc9cdeb57b02890a4e : SUCCESS in 1h 27m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/12457eb9b7cc4f569ae2a5eb1fe37682 : FAILURE in 21m 18s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1f9e3fe659b54445b577c90d1ae50304 : FAILURE in 24m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3d72a66c76014bfa8aef5ddd6c821b49 : SUCCESS in 2h 07m 44s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97f7d04ed6c043078fac51bd1c30f234 : SUCCESS in 2h 00m 06s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d511727f9e647ec80ce41b31899b89e : SUCCESS in 2h 01m 42s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cefc0fe19eef423a9c791c78fb001d09 : FAILURE in 14m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c287eb96c90a46c080657bf2b7977bf0 : FAILURE in 27m 03s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ef4bc813bec436abee34c2cd0fc226b : SUCCESS in 1h 44m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec920fbeca61470abeb78b925bcd9742 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8df7144b521b4fa4b4b242bbdb707991 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":8},{"id":"6ae772413c6973217f09ff0093b0da725045fa5d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-03 11:00:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8},{"id":"2b87599767ff9962a0a8002d7a6efd1bd256fadf","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 21:21:59.000000000","message":"Restored","accounts_in_message":[],"_revision_number":8},{"id":"8b6b621e347101424a848abaa17367b78340677f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 21:24:01.000000000","message":"Topic osa/bindep removed","accounts_in_message":[],"_revision_number":8},{"id":"fe5797a54d69e6714c2cafe7e64ab45a725ecc39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 21:26:49.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"2da58ee61049bb945ad780ee419905ff41e020c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 21:27:18.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"89dac11211259cc7f46c705035af88352cedebb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 21:28:57.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"1089b2b3647f1a22367b0c638e5af1d4755e4243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 22:22:59.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"5efacb323db90e301cd20c0823d099c326558afd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-08 23:02:40.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"49d805de63514f50efd2acdba9e013a2a89c6ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 01:31:46.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4e0c0895e644f2b81b97adc77d080e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4f624d3943f748149503d7b388e2a45d : SUCCESS in 11m 40s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/35515f56167f4f87841d1ba8b52bf14a : SUCCESS in 5m 20s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ea1f8d01f2a646768fdb26b2a7582032 : SUCCESS in 4m 44s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a1b31a4a7317418e8182a8254442ce38 : SUCCESS in 9m 43s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1578f9ca2f804bc3a9472726e8de4e7c : SUCCESS in 11m 26s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/0422b588bc2a48c791734f79edcaed04 : SUCCESS in 7m 30s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/2735ecbfd1134f46bda2735c8512c5e8 : SUCCESS in 7m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/8c5b8e0857634ae5962d76bfc3312580 : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/463d2876f6374024ad6a8ef051e82bac : SUCCESS in 1h 23m 31s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/7b8eda1cf3354d2c9f719c0869a5e9d6 : SUCCESS in 1h 02m 06s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/995392103c4c473499f2b9772d174e4d : SUCCESS in 57m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/087244b1b40f400e8423fd8cc4ceed85 : SUCCESS in 54m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e49de875a3904a7599ad4603b321609a : SUCCESS in 1h 05m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d54b904aa69e4aa4b2c263cdd8a2c4ac : SUCCESS in 1h 18m 04s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c2a2f20c56634d6a9e691f636df3ed1a : SUCCESS in 1h 01m 01s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d7e7f54184b14c46915101d00d1322ee : SUCCESS in 1h 47m 53s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cbc1dac8357b471e9e529831f49fefbe : SUCCESS in 2h 15m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb05a3b196fa4d4c9ab106bb436105b2 : SUCCESS in 2h 14m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3bf9328a55d4427c95232f27fbb95b7d : SUCCESS in 1h 51m 44s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de3878fef10a482ebc79fdb3e9432293 : SUCCESS in 2h 06m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8ddec5a4d67e4c5e8a5711fb2d361fc2 : SUCCESS in 1h 10m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a9d4c840322a454bbc58422e948c27c7 : FAILURE in 11m 01s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c02ef366aa864c2fbfe4b5645ec87f81 : SUCCESS in 2h 05m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e0d2265e09540578ab72d56357ad191 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0686cc44a5a242819535192373f34a02 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":13},{"id":"eb6d2ec7e7523980244427aa6d9717db9c499837","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-10 19:43:00.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"b301a700db7d4c079bf23d276b9b8c2e9b3f887b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 22:14:01.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9feb256e48534192b35d4839c86d8b88\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1663328bca0477a958288ad3cdacb9d : SUCCESS in 8m 35s\n- openstack-ansible-tox-molecule-debian-bookworm https://zuul.opendev.org/t/openstack/build/6106112f47a1477d8ba7f2759b2c5720 : SUCCESS in 6m 21s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/ad86557e396b4cf48ab565a23fe8be1a : SUCCESS in 7m 56s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b6fef44c922426aa963b3f72389c77b : SUCCESS in 7m 57s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d1f35abdedea420bba7a7baf98b1d56d : SUCCESS in 5m 15s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/c40f2dfe2baa499ca38b11bb1839f7a5 : SUCCESS in 7m 21s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8435e8e0f8974398af019ff976fb0b0f : SUCCESS in 6m 54s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3b7f304e10d64afda1f59800174efa59 : SUCCESS in 1h 11m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/0e1f790535a9437796d05b4bedc2050d : SUCCESS in 51m 46s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e6bbb6d8a6de4c719bfd6b9249c65ca1 : SUCCESS in 1h 08m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a22a87c0c96e4e6fac1e303f32edae5c : SUCCESS in 1h 13m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c9275f907479467a8e9675e8bb6d64bb : SUCCESS in 1h 18m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4bb8641408744319b9f5614e9290d2e6 : SUCCESS in 1h 31m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f847456e7a304d829342c27d2e9ce920 : SUCCESS in 47m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3cd27512bec42b6977629517caec5e6 : SUCCESS in 52m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/26dcea7b9b474e2eaebb99b17cdcb96d : SUCCESS in 2h 15m 46s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/30cc58b10c874f4c9e489f214b142ffc : SUCCESS in 2h 15m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ed8db9525184f84b3d32bba7b16e1e4 : SUCCESS in 2h 13m 23s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/94b37dbe431b4c9088a20e74b5bff41c : SUCCESS in 1h 08m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/495371ba171040a1806ae136a4197120 : SUCCESS in 2h 22m 58s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d3de211f391472bb551e0d5374e11e8 : SUCCESS in 1h 21m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7d6535d6e25441c4a6690543a93b219e : SUCCESS in 1h 17m 32s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d1c11749bb74d538e82a7e7fc10a15f : SUCCESS in 58m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a70c5f4a2f84000b155b2c86c7e6406 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15bede8c7f1641f18732eb15c252ba6d : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":14},{"id":"790e0b2d715062150a50dbc1e436a3f6d72a7791","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-06-11 15:36:46.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"f36e2f0940dd227e63ccf42bf0f745ad64121b5d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-25 06:59:01.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"925405e1da1c82b92335f3e4d215eefb8070d48b","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-25 08:31:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":14},{"id":"e9a1e6995a6d74614006fe8216886c27e1e96375","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-25 08:32:15.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"615f0af3c5b969f1fbd09c913926c901563b5d80","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-25 08:32:22.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":15},{"id":"b8b3c3a3f3312bbdf63a6ad3ee1e3af2452f6cea","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-25 08:32:39.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"dd55c8aaf1027577493fb9cf96c7f3aab91df529","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-25 08:36:05.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":15},{"id":"54c3fb7980e1caf51f5ea3431827a81a86055840","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-25 08:36:36.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"305d7e5a18983b872b57041a59e394c0afc5cc32","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-25 08:36:39.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":16},{"id":"a378597f7b0339aa2061c6995f44ca747db210bf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-25 08:50:19.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"97efde3a74ca38229ade2ae7e6fb14fbf514c1f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:03:39.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50888d60eec84a32ab66d29689c6aa3e\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b05f738d58b4b8c80e1eacfd35ae248 : SUCCESS in 9m 02s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/562d5260db624c17b71c05aa3c56033a : SUCCESS in 8m 03s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b8c49374332b40048313dab35a72a2a3 : FAILURE in 4m 13s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/f493374be3204e978f6f4d63cfcfc56b : SUCCESS in 7m 58s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/571eea81574c4064847bc293cc06c716 : SUCCESS in 7m 16s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9b51a6a6aaba4d5084716b2d163ea248 : FAILURE in 5m 27s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3ee9d7b046614799896db353377a6268 : SUCCESS in 52m 51s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/ebc70cf74238465c9379c4fa9d78d690 : SUCCESS in 1h 27m 28s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/941fb7bc6ecc40249b7e2c85f94f30e5 : SUCCESS in 1h 22m 14s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/87a8db8638ea4313a5e86b7d2d688433 : SUCCESS in 1h 39m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8c45978cc02a45e0836720efe6456674 : SUCCESS in 1h 30m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d0f6c0f3dd964daf86e48426eba6c6c5 : SUCCESS in 1h 26m 58s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ac6a1633c164b248f0328e72b21205a : SUCCESS in 55m 02s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/61cb6aa5b88c4feba51ae530f7deeb33 : SUCCESS in 2h 23m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aec46300fd1d49458b29f06a90580f75 : SUCCESS in 1h 58m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/042c71e38a0248879ee8cdb8b9ee28ac : SUCCESS in 2h 15m 05s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/385f638e4dbd49578d7548207cba0ca1 : SUCCESS in 2h 17m 23s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/73dcdb17500a46ff9ec33104b37d7136 : SUCCESS in 1h 13m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c940f80785674283b8440ced54f214e8 : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7e5320efd64c4b50b781ddddeeeabc79 : SUCCESS in 1h 12m 17s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5b629f044f7d4ddba96622b6f92e51ee : SUCCESS in 2h 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee20c1da0ee54429bf08314907f5d9c7 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db0d8bd980d347b7928fc5b11e76c619 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":16},{"id":"33c2e810df1dc8854348dbf96985f150bf48a776","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-25 13:17:39.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"fa9a7c1184b9df2b75ff09640a43b93acf565608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 15:56:31.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa76de90cc87423f832002536ec81ef8\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f055ed57c6af454d951e6eaa632f822b : SUCCESS in 11m 10s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/d06514f476064869be3c50446540fe6f : SUCCESS in 9m 10s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2093e38070cf4844840115ee4960b899 : SUCCESS in 10m 23s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/becc32c0d61a4568b349d163792d3004 : SUCCESS in 4m 46s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/22095a138f5d4a689a0103c587d09ccc : SUCCESS in 7m 18s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c015388fec464ba488f9cfa233f72c8c : FAILURE in 5m 08s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/9ed0f713b576407a9bc7a72355555eec : SUCCESS in 1h 37m 46s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b4355f2a929d4ae19ef8512cae96fb7f : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c95eab1a2cc44b79846c7ad76715d1f4 : SUCCESS in 1h 20m 05s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/49b4d32136ed49e6a26fd66d4fc7f3da : SUCCESS in 56m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7f1a4bec78f49928199fbdc0c5b8b1a : SUCCESS in 1h 32m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/9f1cf62fd754480abc1a3b84a96623a2 : SUCCESS in 1h 31m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a546a440ca114755a6d7ceab51aaa678 : SUCCESS in 1h 46m 43s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9f20920255924b209df09e084e6f8340 : SUCCESS in 1h 50m 45s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4a8e030a26d04a9fa283f5307d3951dd : SUCCESS in 2h 25m 14s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a32e1bf81e84be0b25940801d7b7755 : SUCCESS in 2h 23m 19s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6336e8ea80bd475184bd875c1d2279b9 : SUCCESS in 2h 12m 48s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/792d3f2747e045789d3dc92e83570ab6 : SUCCESS in 2h 27m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a701ebac53d74d239682cfe344a30396 : SUCCESS in 1h 22m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/cb7325449a6c4540b1542e4bc86fd0c6 : SUCCESS in 47m 16s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8b2f3aa3db04d0ca37d3da342a529f0 : SUCCESS in 2h 03m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71f8531fc0f482f822e3e3e2e7b0ae2 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9d2388ae8af4d68a18ac50af8c39887 : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":16},{"id":"5ffe15de1290dfa9de7d88c4dd0d6bd37b23b80f","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-07-14 09:47:40.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"49be8acdeb923350bc105e2f6203b0552e96d3e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 09:51:10.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"f3155f89599bbc50f1c4d9873bbfc36ff6d06993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 12:05:15.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb35515d649f481ea5a9454bcd0fa188\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb28b123f90545db8c6f7ed0128daba3 : SUCCESS in 14m 19s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/78e13f613f004f3da1dd025dcb4e2015 : SUCCESS in 9m 50s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b2859d9c15ac47f18c48b95fe9a81e91 : SUCCESS in 10m 20s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/3d209b2b350043ad8793253848cc7e0a : SUCCESS in 10m 26s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/3588cdc6cccd48a5aaef5f2af84c8b6d : SUCCESS in 7m 52s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/9cb5d5862d294dfa993d43f43e137a99 : FAILURE in 28m 45s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/89782acaae5046219bd72d478eca9d0f : SUCCESS in 1h 32m 03s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e4849fecc2b1435da4a804f973c1c5bc : SUCCESS in 1h 15m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/99cf1175b68a4997a955bdad7b2ded80 : SUCCESS in 1h 10m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0c7779999fe541e181cca630a6db501d : SUCCESS in 1h 19m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/2eee861ee8574c2fa90e428efaf4c1b7 : SUCCESS in 1h 27m 30s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2bd0a573fb28425ea33c106342c4b720 : SUCCESS in 1h 31m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0eaff86737264f2aa014d226a636b666 : SUCCESS in 1h 20m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/d2205883c7eb4cb5b8803ae9934a3594 : SUCCESS in 1h 04m 41s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/046e1f71d91e4ef394a4dec6eab12734 : SUCCESS in 2h 07m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a9a04b1910648a8bb0fcda6f3b8ac73 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d526d3ffd4c46ff90b8326f4c2bc7c1 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":16},{"id":"4d26e00793068d9d538cdb3a472af0f215a520fd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-14 18:07:50.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"e8c4125a3f8eea0da70b092a427845a608b298f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 20:38:24.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7e2137048d049ccaa447b7d8c587efb\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5305a72d4c574208b9ec9e016cc60c5e : SUCCESS in 14m 36s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/594ce66125b74a18b1cb17eecefcad3e : SUCCESS in 9m 16s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/57f5c2af17dc4c6c9fb77b31fffbc4bb : SUCCESS in 5m 53s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/887b03da0d1247499f564471e1b94b24 : SUCCESS in 6m 12s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/64a1391429374db4bbe32747a42579de : SUCCESS in 7m 40s\n- openstack-ansible-tox-molecule-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cc4402a082a946608c8f8093b970ce72 : SUCCESS in 6m 28s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/ccb8fb1acc25432990692e1e8212e646 : SUCCESS in 2h 12m 41s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/a1498a0d6b7d4be6a5a9a1a222b8318d : SUCCESS in 1h 28m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ff9cc93b4cd74832abcee04bf54ca2a3 : SUCCESS in 53m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e2513609fe474adfa63ab2dbabeb7a9c : SUCCESS in 1h 41m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b0c5caef54846ffbdc113c0609e1d72 : SUCCESS in 1h 47m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/de480ba957b44490a35f3cb1e8e3dc4c : SUCCESS in 1h 33m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de1e93720461458dba343475cce46736 : SUCCESS in 55m 18s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/24a067012be34ebf8edce9eea3cb9cb7 : SUCCESS in 1h 18m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d901e50f8604118b25ac671e0b0ceb9 : SUCCESS in 2h 22m 57s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/63026e02ff654c6ab0c462c67c347412 : SUCCESS in 1h 53m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7afe6ee1ca5f40e594da06e2c96669c7 : SUCCESS in 1h 26m 54s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/1dd4378e332f47e3bc7427cd815c558e : SUCCESS in 51m 23s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8391ff2ce9a4240bf1f86ebf8109828 : SUCCESS in 2h 02m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17cf314e8de64916a0b6e1bf91ffb59c : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8327036ae48c4fdbb3252c1e4f3e743a : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":16},{"id":"800e8eb75ceb74f63ad9a6e28b8c3339ce241093","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 20:41:03.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"a4a93ac8f25a83fa7fc00fd70f23e16e6e170558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 23:09:28.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a0acf619e3948f99c6643b9ef39ce5f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fea7fa390444e659ac5e07ad819e5c4 : SUCCESS in 9m 34s\n- openstack-ansible-tox-molecule-debian-trixie https://zuul.opendev.org/t/openstack/build/888a4c47a9de4e048d5ee834cc6e4127 : SUCCESS in 6m 22s\n- openstack-ansible-tox-molecule-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce6abf30bcaf4743b8ec0f67e421cf2a : SUCCESS in 9m 45s\n- openstack-ansible-tox-molecule-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/e65df7c179cb43999afd7c360a2f1e01 : SUCCESS in 10m 44s\n- openstack-ansible-tox-molecule-centos-10-stream https://zuul.opendev.org/t/openstack/build/f01cd23638484414be0c71b8715266c1 : SUCCESS in 7m 11s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c1f5c894fad346079557b6fcaf333c2e : SUCCESS in 2h 16m 44s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c1f7b36d45c843059952ce22f3a48f6a : SUCCESS in 1h 24m 14s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b7efa0b700e143ccb252ce39b8b3c246 : SUCCESS in 55m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/4c19cde91b484cec82c03105b73db3a3 : SUCCESS in 1h 14m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b839383f66e432a8a5819264407e9e5 : SUCCESS in 1h 32m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/a79561d0b28841b099fe2495dd105d57 : SUCCESS in 1h 30m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/85a075c6d0784a4d8d3345d25177a218 : SUCCESS in 1h 31m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/14fdcce73a2c4fbba782b5aaad10345c : SUCCESS in 1h 20m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/13d1d10d2daa482e8b0b7f3ab92d3868 : SUCCESS in 1h 39m 29s\n- openstack-ansible-upgrade-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c22ce5d641304dbe8470ec0e75443779 : SUCCESS in 2h 09m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/794f1c73519f4c998214ced74c7551e3 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92ffeda4aab94874a8dfa27fe263f03f : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":16},{"id":"68d26136b7f7072baae176f1f49cf60aa3c025b0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 23:09:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"fff3d4cbf6db5b8e209f518985a3a17b76c3f5a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 23:10:24.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba02b0407c864e858e8bf2da68a0daae\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ceb44459ef14626906024e39db00a6c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"748547aeee5da9d2e4371193839fe031170a4d00","revisions":{"63a36f38515fffae16d3a4b93de6d32039ff6bae":{"kind":"REWORK","_number":1,"created":"2025-12-23 23:28:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/1"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"subject":"Remove vars for unsupported OS","message":"Remove vars for unsupported OS\n\nFor RHEL10 rename to RHEL.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/63a36f38515fffae16d3a4b93de6d32039ff6bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/63a36f38515fffae16d3a4b93de6d32039ff6bae"}]},"branch":"refs/heads/master"},"7e4013441ba940d021439b1d18a303fce3fb927b":{"kind":"REWORK","_number":2,"created":"2025-12-23 23:34:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/2"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:34:27.000000000","tz":0},"subject":"Remove vars for unsupported OS","message":"Remove vars for unsupported OS\n\nFor RHEL10 rename to RHEL.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/7e4013441ba940d021439b1d18a303fce3fb927b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/7e4013441ba940d021439b1d18a303fce3fb927b"}]},"branch":"refs/heads/master"},"bbdf43d0f8a6798b02eff0ce7f2a95e4a4b62243":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-12-23 23:35:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/3"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:35:09.000000000","tz":0},"subject":"Remove vars for unsupported OS","message":"Remove vars for unsupported OS\n\nRHEL10 renamed in RHEL, Ubuntu-24 in Ubuntu.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/bbdf43d0f8a6798b02eff0ce7f2a95e4a4b62243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/bbdf43d0f8a6798b02eff0ce7f2a95e4a4b62243"}]},"branch":"refs/heads/master","description":"Edit commit message"},"237a80d4a1a7bde5befbe3067aa5e0529d1d6632":{"kind":"REWORK","_number":4,"created":"2025-12-23 23:39:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/4"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:38:50.000000000","tz":0},"subject":"Remove vars for unsupported OS","message":"Remove vars for unsupported OS\n\nRHEL10 renamed in RHEL, Ubuntu-24 in Ubuntu.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/237a80d4a1a7bde5befbe3067aa5e0529d1d6632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/237a80d4a1a7bde5befbe3067aa5e0529d1d6632"}]},"branch":"refs/heads/master"},"fcca86a3e578a96defa967f653450c2ff5483e66":{"kind":"REWORK","_number":5,"created":"2025-12-24 00:00:50.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/5"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-24 00:00:49.000000000","tz":0},"subject":"Remove vars for unsupported OS","message":"Remove vars for unsupported OS\n\nRHEL10 renamed in RHEL, Ubuntu-24 in Ubuntu.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/fcca86a3e578a96defa967f653450c2ff5483e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/fcca86a3e578a96defa967f653450c2ff5483e66"}]},"branch":"refs/heads/master"},"6b8afb0bc5e68967155b8e024b527711605904db":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-12-24 00:03:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/6"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-24 00:03:25.000000000","tz":0},"subject":"Remove vars for unsupported OS","message":"Remove vars for unsupported OS\n\nRHEL10 renamed in RHEL, Ubuntu 24 in Ubuntu, Debian 12/13 in Debian.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/6b8afb0bc5e68967155b8e024b527711605904db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/6b8afb0bc5e68967155b8e024b527711605904db"}]},"branch":"refs/heads/master","description":"Edit commit message"},"77181910a6c137e7640a6cc5e7e66969d7e12955":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-12-24 00:21:03.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/7"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-24 00:21:03.000000000","tz":0},"subject":"Remove vars for unsupported OS","message":"Remove vars for unsupported OS\n\nRHEL10 renamed in RHEL, Ubuntu 24 renamed in Ubuntu, Debian 12/13 in Debian.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/77181910a6c137e7640a6cc5e7e66969d7e12955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/77181910a6c137e7640a6cc5e7e66969d7e12955"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e7d6fd4ad5375ec2b27ec83c4b767b14d1d47875":{"kind":"REWORK","_number":8,"created":"2026-01-12 16:58:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/8"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 16:58:05.000000000","tz":0},"subject":"Update URL in bindep.txt","message":"Update URL in bindep.txt\n\ntrivial fix\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/e7d6fd4ad5375ec2b27ec83c4b767b14d1d47875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/e7d6fd4ad5375ec2b27ec83c4b767b14d1d47875"}]},"branch":"refs/heads/master"},"137aca42e4b22dba7c9f7b2d712b764778f10cba":{"kind":"REWORK","_number":9,"created":"2026-06-08 21:26:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/9"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-08 21:26:07.000000000","tz":0},"subject":"Update URL in bindep.txt","message":"Update URL in bindep.txt\n\ntrivial fix\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/137aca42e4b22dba7c9f7b2d712b764778f10cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/137aca42e4b22dba7c9f7b2d712b764778f10cba"}]},"branch":"refs/heads/master"},"b1555543a45cdae2f038c10fb15461fb115fd596":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2026-06-08 21:27:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/10"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-08 21:27:18.000000000","tz":0},"subject":"Remove support for RHEL 7, 8 and SUSE 42","message":"Remove support for RHEL 7, 8 and SUSE 42\n\nAll operation systems already EOL.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/b1555543a45cdae2f038c10fb15461fb115fd596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/b1555543a45cdae2f038c10fb15461fb115fd596"}]},"branch":"refs/heads/master","description":"Edit commit message"},"267a8c2afab7e29ab7fa4ce0269211ad4f8a420b":{"kind":"REWORK","_number":11,"created":"2026-06-08 21:28:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/11"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-08 21:28:55.000000000","tz":0},"subject":"Remove support for RHEL 7, 8 and SUSE 42","message":"Remove support for RHEL 7, 8 and SUSE 42\n\nAll operation systems already EOL.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/267a8c2afab7e29ab7fa4ce0269211ad4f8a420b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/267a8c2afab7e29ab7fa4ce0269211ad4f8a420b"}]},"branch":"refs/heads/master"},"9b0e284699277e53029e5b5e6791405d2b5311b9":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2026-06-08 22:22:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/12"}}},"commit":{"parents":[{"commit":"d6ffa1f2973c1fb246ddc236419b930d95156a41","subject":"Merge \"Remove unused directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/d6ffa1f2973c1fb246ddc236419b930d95156a41"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-08 22:22:59.000000000","tz":0},"subject":"Remove support for RHEL 7, 8 and SUSE 42","message":"Remove support for RHEL 7, 8 and SUSE 42\n\nOperating systems already EOL.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/9b0e284699277e53029e5b5e6791405d2b5311b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/9b0e284699277e53029e5b5e6791405d2b5311b9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4b29cec6f3718f7a3589de14696ba30dd7ff4f45":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-06-08 23:02:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/13"}}},"commit":{"parents":[{"commit":"e9d6dc3388bcc5ddd65a9bfae6b28c7fbc0e6dc6","subject":"Add support for Ubuntu 26.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/e9d6dc3388bcc5ddd65a9bfae6b28c7fbc0e6dc6"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-08 23:02:40.000000000","tz":0},"subject":"Remove support for RHEL 7, 8 and SUSE 42","message":"Remove support for RHEL 7, 8 and SUSE 42\n\nOperating systems already EOL.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/4b29cec6f3718f7a3589de14696ba30dd7ff4f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/4b29cec6f3718f7a3589de14696ba30dd7ff4f45"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d6ffa1f2973c1fb246ddc236419b930d95156a41","ours":"9b0e284699277e53029e5b5e6791405d2b5311b9","theirs":"e9d6dc3388bcc5ddd65a9bfae6b28c7fbc0e6dc6","merge_strategy":"recursive","contains_conflicts":false}},"cdac23cb258a3005b830d66d79f21580d637bc99":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-06-10 19:43:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/14"}}},"commit":{"parents":[{"commit":"44167206fcb011996de73d90a3fced3e6b9e76e2","subject":"Drop support for Debian 11 and earlier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/44167206fcb011996de73d90a3fced3e6b9e76e2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-10 19:43:00.000000000","tz":0},"subject":"Remove support for RHEL 7, 8 and SUSE 42","message":"Remove support for RHEL 7, 8 and SUSE 42\n\nOperating systems already EOL.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/cdac23cb258a3005b830d66d79f21580d637bc99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/cdac23cb258a3005b830d66d79f21580d637bc99"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e9d6dc3388bcc5ddd65a9bfae6b28c7fbc0e6dc6","ours":"4b29cec6f3718f7a3589de14696ba30dd7ff4f45","theirs":"44167206fcb011996de73d90a3fced3e6b9e76e2","merge_strategy":"recursive","contains_conflicts":false}},"6e219351afb7958e1ecc0cd27e3add932bc1a365":{"kind":"REWORK","_number":15,"created":"2026-06-25 08:32:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/15"}}},"commit":{"parents":[{"commit":"44167206fcb011996de73d90a3fced3e6b9e76e2","subject":"Drop support for Debian 11 and earlier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/44167206fcb011996de73d90a3fced3e6b9e76e2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-25 08:32:14.000000000","tz":0},"subject":"Remove support for RHEL 7 and SUSE 42","message":"Remove support for RHEL 7 and SUSE 42\n\nOperating systems already EOL.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/6e219351afb7958e1ecc0cd27e3add932bc1a365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/6e219351afb7958e1ecc0cd27e3add932bc1a365"}]},"branch":"refs/heads/master"},"748547aeee5da9d2e4371193839fe031170a4d00":{"kind":"REWORK","_number":16,"created":"2026-06-25 08:36:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/45/971745/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ansible-role-systemd_networkd","ref":"refs/changes/45/971745/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ansible-role-systemd_networkd refs/changes/45/971745/16"}}},"commit":{"parents":[{"commit":"44167206fcb011996de73d90a3fced3e6b9e76e2","subject":"Drop support for Debian 11 and earlier","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/44167206fcb011996de73d90a3fced3e6b9e76e2"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-12-23 23:28:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-25 08:36:34.000000000","tz":0},"subject":"Remove support for RHEL 7 and SUSE 42","message":"Remove support for RHEL 7 and SUSE 42\n\nOperating systems already EOL.\n\nChange-Id: Icb5b53ef85f5a9b8a4a59c554739129d25a0bb29\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/748547aeee5da9d2e4371193839fe031170a4d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ansible-role-systemd_networkd/commit/748547aeee5da9d2e4371193839fe031170a4d00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
