)]}'
{"id":"openstack%2Faodh~1001500","triplet_id":"openstack%2Faodh~master~Ia0ffb21b3928a1677dd97c0e26686f53fd9f9bef","project":"openstack/aodh","branch":"master","topic":"bug/2161276","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-19 15:30:08.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":["CVE-2026-76878","OSSA-2026-036"],"change_id":"Ia0ffb21b3928a1677dd97c0e26686f53fd9f9bef","subject":"Fix all_projects\u003dfalse bypass project scope","status":"MERGED","created":"2026-08-19 14:07:25.000000000","updated":"2026-08-20 06:15:02.000000000","submitted":"2026-08-19 16:29:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001500-bug/2161276","meta_rev_id":"90049d14e91d8dacb917861e5d632ac813ed264e","_number":1001500,"virtual_id_number":1001500,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-19 16:29:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2026-08-19 14:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2026-08-19 15:30:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-19 14:09:42.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2026-08-19 15:29:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca2a34d693d5143a716f65fd476f61a1d7517fa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-19 14:07:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0857c63b5072777bff06712a67fcbd341a107029","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-08-19 14:09:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d38ea28bb85b9e16e5a818eddbb7f359904594f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 15:01:49.000000000","message":"Hashtag added: OSSA-2026-036","accounts_in_message":[],"_revision_number":1},{"id":"9eff3c99f8a453be1596206a6a831f871feb4f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 15:29:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1104cc2aa6f4e0e8d8d4c31251140c1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e42922a1d5c64b308c65476584e5772a : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/084c19a8f4b1409a8d2e06032ba403e6 : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8e6462e16d054fe6b2c7b76d6f2b7bf4 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a25fa9f89ef44f2889704c4c0141679a : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c149d729193c41389fed84c75542d4e2 : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8931327d07849f1a2d99b5d0788ba83 : SUCCESS in 4m 48s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/d1b6fff154374dd1bad4144bff3f091d : SUCCESS in 1h 16m 04s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/fb256776592c428780555444dc57d899 : SUCCESS in 47m 38s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/29378cfafa1b4198a241fb48c35b8379 : SUCCESS in 30m 57s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6404fa405f1642a88a34fdc6fecd6df8 : SUCCESS in 53m 43s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/462fb683016f44b88fbcd803c7b268a1 : SUCCESS in 47m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a0fc340d49127c42d7568c681429942253d21b8e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-19 15:30:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17813e6e49383e299c30374cc0218f8d70d08449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 15:30:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"472c10e1d13ad40a9b6ca5812c1a3d212fd58ca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:29:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cac44f713c364155aba5302538964c71\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8366c30e65c145b3b9c85241f9d06ae2 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/faaac9b85ff34b629607d2ea19c02eb2 : SUCCESS in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d5a0537a79f14f63a63caa37d9eb54e9 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa81eda447a4aa69fd9d11524ade245 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1022c3f2c8c24f20808d73bb89152c0e : SUCCESS in 6m 18s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/aa6943366121461c89fc9490af18c7b4 : SUCCESS in 57m 33s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f59d679f4c6c4c409a42f73cd45d80db : SUCCESS in 44m 31s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/ca7905e50e584816a9c452d403029ba7 : SUCCESS in 28m 02s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/aaa4046e3d9642b4844827cb3d804fa9 : SUCCESS in 42m 37s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/331777a2fc2948e29f7f53bc226aa599 : SUCCESS in 45m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7b8ad2798c520c54229c78ae313de1900585a7ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:29:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"25cf5489a4c0d009457e8b30eb6fb9b0179b03fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:31:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4de72bf4a009442a97faf7e87f11cebf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69cd028de88e4c4ba4cee6a7a5be94bc : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cae85265c6144a9faa364f0870487441 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"90049d14e91d8dacb917861e5d632ac813ed264e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-20 06:15:02.000000000","message":"Hashtag added: CVE-2026-76878","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67dc527a291c75535258fc0d2c5381b261453d64","revisions":{"67dc527a291c75535258fc0d2c5381b261453d64":{"kind":"REWORK","_number":1,"created":"2026-08-19 14:07:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/00/1001500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/00/1001500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/00/1001500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/00/1001500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/00/1001500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/00/1001500/1"}}},"commit":{"parents":[{"commit":"aa8cdb178c9a51a74d916ee14cacb26be0970966","subject":"Merge \"Replace deprecated datetime.datetime.utcnow\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/aa8cdb178c9a51a74d916ee14cacb26be0970966"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-19 03:34:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-19 03:34:39.000000000","tz":540},"subject":"Fix all_projects\u003dfalse bypass project scope","message":"Fix all_projects\u003dfalse bypass project scope\n\nModify how all_projects is handled in queries using the alarms endpoint.\nPreviously using all_project\u003dfalse would allow skipping a branch of the\ncode, which scopes requests to the current project. With the new code,\nwhen all_projects\u003dfalse is present, the project scoping is done as\noriginally intended.\n\nCloses-Bug: #2161276\nCo-Authoered-By: Jaromir Wysoglad \u003cjwysogla@redhat.com\u003e\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nChange-Id: Ia0ffb21b3928a1677dd97c0e26686f53fd9f9bef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/67dc527a291c75535258fc0d2c5381b261453d64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/67dc527a291c75535258fc0d2c5381b261453d64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
