)]}'
{"id":"openstack%2Faodh~1001522","triplet_id":"openstack%2Faodh~stable%2F2025.1~I10ead0873dd59fca602d2cae958e193020eae2fe","project":"openstack/aodh","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-19 16:23:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I10ead0873dd59fca602d2cae958e193020eae2fe","subject":"stable-only: Disable failing tests","status":"MERGED","created":"2026-08-19 14:58:59.000000000","updated":"2026-08-19 17:28:59.000000000","submitted":"2026-08-19 17:27:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1001522","meta_rev_id":"db06b86f2d24ead51b93cb708e029a6e70939f3d","_number":1001522,"virtual_id_number":1001522,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-19 17:27:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2026-08-19 15:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2026-08-19 16:23:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-19 15:40:28.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2026-08-19 16:07:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a193f29228224c973210c075958e5728bffebd8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-19 14:58:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ad9e14e39cafde9cfeb1089bb024e457a7226c1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-08-19 15:40:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a40f30d290bce573178797759cf93e192d75de7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:07:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27dcc2676a2a49e9aa6ca83a968d81fc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e734813b695446dab311926a8d116b5 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc348b348ac445b8a3a4fb3535fafe4f : SUCCESS in 6m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f09a96913444d789244d03f57ce223a : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d30a6a00e044f55b3ed51d64198b215 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d7ace23e29449168abe0779051e8e85 : SUCCESS in 6m 34s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8588fb9f6e1543d288b2da5a07c1a350 : SUCCESS in 44m 51s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/344eb4f70b2c492680fd609c1284a354 : SUCCESS in 40m 01s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4d8e84ebbb1347a1a77270b4997aa746 : SUCCESS in 52m 07s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d27845c909df4f888852ddae341b9edd : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0d44f4a6382c1d7a426e05eebe1d4ff8fcc50166","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-19 16:23:14.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d30897d0894c7f4b8016074e3be987f9eee0b4c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:24:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2a07013a2094743fc7a66c3b69bdd76c8319ac02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 17:27:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d751b8098e9c461e97578d07cfd94553\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8004100ad2d8408992f20ee78d262d1d : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed22c045dce543cd861fad5fa5dd5e9c : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93be46c4d33f4f0c99a89df74835b8b4 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07303297f72d4134bf42163c65e45217 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56007c05bf024a5f93cedcadf3cc11af : SUCCESS in 4m 27s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d852db870ba64b42bc5b95ac01d2e764 : SUCCESS in 39m 24s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/2a760abc2e0c46998afdbc598a35c7dc : SUCCESS in 28m 57s (non-voting)\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c2939f80d91749a5824e5b2b7e3b20bb : SUCCESS in 33m 39s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/854dc277b30c46e39de6ca8209183ef5 : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e43e1eaf7e238726edf11ff9621f205af77d6e77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 17:27:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"db06b86f2d24ead51b93cb708e029a6e70939f3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 17:28:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0aad037499994ab699b759260dbf9a07\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cec48461269e48fa9231c82c806daecf : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b1b3aaea4f279fcf789a6fdb3e1442f0ef00b39","revisions":{"6b1b3aaea4f279fcf789a6fdb3e1442f0ef00b39":{"kind":"REWORK","_number":1,"created":"2026-08-19 14:58:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/1001522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/22/1001522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/1001522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/1001522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/1001522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/22/1001522/1"}}},"commit":{"parents":[{"commit":"f73d4ed13642515db7404444ef8fa54a5ebc0c0c","subject":"stable-only: Disable config file generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f73d4ed13642515db7404444ef8fa54a5ebc0c0c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-19 14:57:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-19 14:57:47.000000000","tz":540},"subject":"stable-only: Disable failing tests","message":"stable-only: Disable failing tests\n\nJammy jobs are failing due to removed Python 3.10 support. Also CentOS\nStream 9 job is broken due to problem with library installation.\n\nDisable these jobs to unblock CI so that we can merge an urgent\nsecurity fix.\n\nChange-Id: I10ead0873dd59fca602d2cae958e193020eae2fe\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6b1b3aaea4f279fcf789a6fdb3e1442f0ef00b39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6b1b3aaea4f279fcf789a6fdb3e1442f0ef00b39"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
