)]}'
{"id":"openstack%2Faodh~1002010","triplet_id":"openstack%2Faodh~master~I3c8cbdb86d9978cc83f83d4ad51049b52e5be4ea","project":"openstack/aodh","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-31 15:31:39.000000000","reason":"Change was submitted"}},"hashtags":["remove-enforce-scope-flag"],"change_id":"I3c8cbdb86d9978cc83f83d4ad51049b52e5be4ea","subject":"Remove disabling the scope enforcement","status":"MERGED","created":"2026-08-24 05:30:16.000000000","updated":"2026-08-31 15:32:53.000000000","submitted":"2026-08-31 15:31:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1002010-secure-rbac","meta_rev_id":"15b760057b935042c39f9f5aaf37210dad68f159","_number":1002010,"virtual_id_number":1002010,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-31 15:31:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2026-08-24 08:55:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-31 13:47:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-31 13:47:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-24 06:24:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-24 08:55:49.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2026-08-31 13:47:14.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"}],"messages":[{"id":"5f3faa89d6da03ba4ea0ba82b4d400b256163ebc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 05:30:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7c819c2cda68669af9873c3795354160029be10","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-24 05:30:26.000000000","message":"Hashtag added: remove-enforce-scope-flag","accounts_in_message":[],"_revision_number":1},{"id":"ee5086c84b00f3e79cd78820d60425c7ea887b90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 06:24:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea52881cfbff4b77b1eeeddab628b24b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900c22732b6144b88bcb6d813bc40ca5 : SUCCESS in 3m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/483472412031440da266bf2370781983 : SUCCESS in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/474336855b754901b667b96c5d4197d4 : SUCCESS in 2m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8fdce6e2ff36424e935a4dbd63ad0a16 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b11209b91bfc4687a4548c6871c83884 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b0f262a50604485a6a5a0232ebf6a69 : SUCCESS in 5m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/15caa1d65cb34fa1b49da73914d20c89 : SUCCESS in 2m 42s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/ad6b9b4714004f9d8db6ef8c412fb842 : SUCCESS in 35m 00s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cc6840a10e5f4dc69b75a6a7aa454db5 : SUCCESS in 48m 14s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/d2ddf41c597d45119faa9fac6950a154 : SUCCESS in 42m 50s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/205d63b12df94bda9c5cfff7b4de26e2 : SUCCESS in 40m 47s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/9656febc54ca451c89304cdb448382c8 : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1489dc26b19a35607f6407d2cbb43f0866d41d99","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2026-08-24 08:55:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc3bee2a6b27b15f35607abd91186ec480213800","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2026-08-31 13:47:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5681ddc1dc1eb1080b37263c5f0f5b3ece37661d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 13:47:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1de1f349217df716bf35c596d8740eede81d51e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 15:31:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c45b9d5ea77d46458e11bf54c5b5f240\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be5c2a0ab204834aab96246be4091dc : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23c39b3b60544ddd971de400ceb7cc05 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cb17fa9fe750492382ceb59692769892 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a013314b94c143f4baba8a4a1b08de0d : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0b0d3babc2443288be036663afd1da3 : SUCCESS in 6m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c4ff5453ba8347e093b4f21edd010adb : SUCCESS in 4m 49s\n- grenade-aodh https://zuul.opendev.org/t/openstack/build/0d977fd2b07d4d418b324642f7ff9544 : SUCCESS in 1h 08m 49s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2f2cf8a860ee4b8a85e01c20cf0b66d7 : SUCCESS in 37m 37s\n- telemetry-dsvm-integration-aetos https://zuul.opendev.org/t/openstack/build/bb0cc07a5f8e4caebb13f59c9a6d844c : SUCCESS in 44m 41s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/84a19dff864747e38181f883c4706d33 : SUCCESS in 32m 41s\n- telemetry-dsvm-integration-aetos-ipv6-only https://zuul.opendev.org/t/openstack/build/aec90cf1845440ff8a5e68b840c2047e : SUCCESS in 43m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5e1d2b7ff784b4184b9a4a4fc7b667a251c905d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 15:31:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"15b760057b935042c39f9f5aaf37210dad68f159","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 15:32:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cfaf6b6247f458cae0e67e7472b6e53\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37fcd0f5c16047f8b8150451c8c5c543 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/929594e0ae9f44ecb536390d3ab785a2 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3a37f529e8bd219c8462fa101ad0242508294683","revisions":{"3a37f529e8bd219c8462fa101ad0242508294683":{"kind":"REWORK","_number":1,"created":"2026-08-24 05:30:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/10/1002010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/10/1002010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/10/1002010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/10/1002010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/10/1002010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/10/1002010/1"}}},"commit":{"parents":[{"commit":"a52eaa00527eb71b50d5b5301bad8dcab84fb3c2","subject":"Merge \"Do not skip tempest tests for requirements update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a52eaa00527eb71b50d5b5301bad8dcab84fb3c2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 05:29:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-24 05:29:49.000000000","tz":540},"subject":"Remove disabling the scope enforcement","message":"Remove disabling the scope enforcement\n\nRBAC scope enforcement was controlled by the oslo.policy\ndeprecated config option enforce_scope which has been removed\nin oslo.policy 6.0.0. That removes the ability to disable\nthe scope enforcement which is the end goal of SRBAC.\n\n- https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#id5\n\nAdding upgrade impact in releasenotes with olso.policy version\nbump in requirements.txt.\n\nDepends-On: https://review.opendev.org/c/openstack/oslo.policy/+/986475\nDepends-On: https://review.opendev.org/c/openstack/releases/+/994789\n\nCo-Authored-By: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nChange-Id: I3c8cbdb86d9978cc83f83d4ad51049b52e5be4ea\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3a37f529e8bd219c8462fa101ad0242508294683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3a37f529e8bd219c8462fa101ad0242508294683"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
