)]}'
{"id":"openstack%2Faodh~627229","triplet_id":"openstack%2Faodh~master~I91544cee4c610ec5cb23f73f63efbd90da66b90a","project":"openstack/aodh","branch":"master","topic":"bug/1754212","attention_set":{},"removed_from_attention_set":{"17109":{"account":{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},"last_update":"2025-08-12 16:13:23.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-08-12 16:13:23.000000000","reason":"Change was submitted"},"14107":{"account":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"last_update":"2025-08-12 16:13:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I91544cee4c610ec5cb23f73f63efbd90da66b90a","subject":"Replace independent aodh-api service by Apache web server","status":"MERGED","created":"2018-12-25 07:04:30.000000000","updated":"2025-08-12 16:16:07.000000000","submitted":"2025-08-12 16:13:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"627229-bug/1754212","meta_rev_id":"bc38d082bc1e6bc86044b36031437ff07a16cb15","_number":627229,"virtual_id_number":627229,"owner":{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-12 16:13:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},{"value":2,"date":"2025-08-12 14:59:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-04 09:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},{"value":1,"date":"2025-08-12 14:59:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-25 07:53:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-20 09:03:41.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2025-08-04 09:02:15.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"state":"REVIEWER"},{"updated":"2025-08-12 14:59:05.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"11a37ca69261a43eaab67bd8e4ca7a8b26df3311","author":{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},"date":"2018-12-25 07:04:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cacea81420e0c0b631de5f8760f30cefd9e8b63b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-25 07:53:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/29/627229/1/check/openstack-tox-pep8/0fc3da2/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/29/627229/1/check/openstack-tox-docs/c991ba0/html/ : SUCCESS in 3m 31s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/29/627229/1/check/aodh-dsvm-tempest-plugin-mysql/eafcef9/ : SUCCESS in 35m 17s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/29/627229/1/check/aodh-dsvm-tempest-plugin-postgresql/5eae69d/ : SUCCESS in 36m 25s\n- telemetry-dsvm-integration http://logs.openstack.org/29/627229/1/check/telemetry-dsvm-integration/bbecf24/ : SUCCESS in 42m 20s","accounts_in_message":[],"_revision_number":1},{"id":"616315067563255befb3488f0d0ad30e5c9bf935","author":{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},"date":"2019-05-20 03:14:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"882ce668d08702b581e42c24a180103f5dca2eaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 03:57:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/29/627229/1/check/openstack-tox-pep8/4cf4cb2/ : SUCCESS in 3m 55s\n- openstack-tox-docs http://logs.openstack.org/29/627229/1/check/openstack-tox-docs/52de1ae/html/ : SUCCESS in 3m 17s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/29/627229/1/check/aodh-dsvm-tempest-plugin-mysql/4ccb8d1/ : SUCCESS in 41m 53s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/29/627229/1/check/aodh-dsvm-tempest-plugin-postgresql/d7bcb26/ : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b937437666551627ac88e7e92ae480fdc8c0bae2","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-05-20 09:03:41.000000000","message":"Patch Set 1: Code-Review-1\n\naccording the configure step, there no config for aodh api run under apache. so I think this docs don\u0027t mean run aodh-api under apache.\nbeside, If you change aodo-api run under apache, please also change install-rdo and install-obs","accounts_in_message":[],"_revision_number":1},{"id":"1c2d4e307a6bcd068cb5836aff2fda6001557769","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 15:59:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b72c41cf3e618026eebb3d8000e2f3edef56b459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 16:33:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad7be59127d94b42a2a2aba2133b31ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a497334ab7b04f0a8e2c37bb978d9b8e : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62bdd67d9d0f473c96aa8b6cd18a1593 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":2},{"id":"c11fe681ffa11c1e1b5e2e3d16e63175e61c1456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-06 16:14:53.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f0493dcb002f69f6b19aaad32fccb31597dfe335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 16:28:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb0fcf9761ff4a10b158e55052871c78\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33e030263f844afbbf5fff7c34c90537 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739af71520dc48f6bdb122fc05cdf240 : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c8f3ffdabed3e530fa9c0d3c00572964f540933b","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2025-08-04 09:02:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9c50911c0068199ba608c20384db6afae063be03","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2025-08-12 14:59:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"27629d235faa624344d7a8380c0b75452857b2f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 14:59:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"92cf463eda5c8859d7b6f0149eb0234b9defdd6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 16:13:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c36a5a50a264f43b2f626b51e89860b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5643c81157244de3b1ddd25cb68c093c : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5c5388b7574cffa84170e4220b1ec7 : SUCCESS in 8m 44s","accounts_in_message":[],"_revision_number":3},{"id":"f1c4c12870a9d7fda76d09625afbd70692979870","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 16:13:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bc38d082bc1e6bc86044b36031437ff07a16cb15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 16:16:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a32a525880146cc9277ddaaa5568af5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f85aa7d7b94346019dcca9e1fe8ba8a9 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ab53622cd0f68facb47e9236699edb2bb51672c0","revisions":{"a8ae1fc2cfb56a8b2afe48910a722f47ac889ea2":{"kind":"REWORK","_number":1,"created":"2018-12-25 07:04:30.000000000","uploader":{"_account_id":17109,"name":"Chiawei Xie","email":"jeffxie@iii.org.tw","username":"dommgifer"},"ref":"refs/changes/29/627229/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/29/627229/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/627229/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/627229/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/627229/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/29/627229/1"}}},"commit":{"parents":[{"commit":"c676eae200c8f075342a57867a9cbf302ce0530b","subject":"remove those copy words occured twice times in newton.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c676eae200c8f075342a57867a9cbf302ce0530b"}]}],"author":{"name":"dommgifer","email":"jeffxie@iii.org.tw","date":"2018-12-25 07:00:16.000000000","tz":480},"committer":{"name":"dommgifer","email":"jeffxie@iii.org.tw","date":"2018-12-25 07:04:28.000000000","tz":480},"subject":"Change service aodh-api restart command","message":"Change service aodh-api restart command\n\nAodh-api is running under apache wsgi,\nso restart aodh-api servier by apahche2.\n\nChange-Id: I91544cee4c610ec5cb23f73f63efbd90da66b90a\nClose-bug: #1754212\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a8ae1fc2cfb56a8b2afe48910a722f47ac889ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a8ae1fc2cfb56a8b2afe48910a722f47ac889ea2"}]},"branch":"refs/heads/master"},"98550b46ae7c818485511cc0de1168f7a4d2fb3c":{"kind":"REWORK","_number":2,"created":"2024-11-22 15:59:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/627229/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/29/627229/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/627229/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/627229/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/627229/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/29/627229/2"}}},"commit":{"parents":[{"commit":"c676eae200c8f075342a57867a9cbf302ce0530b","subject":"remove those copy words occured twice times in newton.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c676eae200c8f075342a57867a9cbf302ce0530b"}]}],"author":{"name":"dommgifer","email":"jeffxie@iii.org.tw","date":"2018-12-25 07:00:16.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 15:58:39.000000000","tz":540},"subject":"Replace independent aodh-api service by Apache web server","message":"Replace independent aodh-api service by Apache web server\n\nThe independent aodh-api service has been removed and it is now run by\napache + mod_wsgi. Update the services mentioned accordingly.\n\nNote:\n - Apache Web server is already installed/enabled when keystone is\n   is installed\n - Guide for OpenSUSE is not updated by this change, because OpenStack\n   packages for OpenSUSE haven\u0027t been for long time.\n\nClose-bug: #1754212\nChange-Id: I91544cee4c610ec5cb23f73f63efbd90da66b90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/98550b46ae7c818485511cc0de1168f7a4d2fb3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/98550b46ae7c818485511cc0de1168f7a4d2fb3c"}]},"branch":"refs/heads/master"},"ab53622cd0f68facb47e9236699edb2bb51672c0":{"kind":"REWORK","_number":3,"created":"2025-01-06 16:14:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/627229/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/29/627229/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/627229/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/627229/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/29/627229/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/29/627229/3"}}},"commit":{"parents":[{"commit":"c676eae200c8f075342a57867a9cbf302ce0530b","subject":"remove those copy words occured twice times in newton.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c676eae200c8f075342a57867a9cbf302ce0530b"}]}],"author":{"name":"dommgifer","email":"jeffxie@iii.org.tw","date":"2018-12-25 07:00:16.000000000","tz":480},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-01-06 16:14:53.000000000","tz":0},"subject":"Replace independent aodh-api service by Apache web server","message":"Replace independent aodh-api service by Apache web server\n\nThe independent aodh-api service has been removed and it is now run by\napache + mod_wsgi. Update the services mentioned accordingly.\n\nNote:\n - Apache Web server is already installed/enabled when keystone is\n   is installed\n - Guide for OpenSUSE is not updated by this change, because OpenStack\n   packages for OpenSUSE haven\u0027t been for long time.\n\nClose-bug: #1754212\nChange-Id: I91544cee4c610ec5cb23f73f63efbd90da66b90a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ab53622cd0f68facb47e9236699edb2bb51672c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ab53622cd0f68facb47e9236699edb2bb51672c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
