)]}'
{"id":"openstack%2Faodh~654314","triplet_id":"openstack%2Faodh~master~I5adbc8490a6ab23d080e60b813febd1334f3f4e0","project":"openstack/aodh","branch":"master","topic":"opendev","hashtags":[],"change_id":"I5adbc8490a6ab23d080e60b813febd1334f3f4e0","subject":"Replace git.openstack.org URLs with opendev.org URLs","status":"MERGED","created":"2019-04-22 06:21:16.000000000","updated":"2019-04-22 08:25:34.000000000","submitted":"2019-04-22 08:25:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654314-1555921534415-76a6aaaf","meta_rev_id":"c7f2f9b0bb1d67647e6a8f054e987ee6f39ebd16","_number":654314,"virtual_id_number":654314,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-04-22 08:25:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-04-22 06:30:44.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-04-22 06:30:44.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 06:30:44.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-04-22 08:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e0a47319a625ecf2a216104af4a7284fe8dec91","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-22 06:21:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16e9e05288ccefe1b9aa501c8d4a68a2f42fc709","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-04-22 06:30:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db254fdf4f39cee91be8b51582db06817f901f0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 07:20:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/654314/1/check/openstack-tox-pep8/0e03661/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/14/654314/1/check/openstack-tox-py27/df61303/ : SUCCESS in 6m 41s\n- openstack-tox-py36 http://logs.openstack.org/14/654314/1/check/openstack-tox-py36/68d939f/ : SUCCESS in 6m 49s\n- openstack-tox-py37 http://logs.openstack.org/14/654314/1/check/openstack-tox-py37/ec20f3d/ : SUCCESS in 8m 18s\n- openstack-tox-docs http://logs.openstack.org/14/654314/1/check/openstack-tox-docs/eed6e29/html/ : SUCCESS in 4m 35s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/14/654314/1/check/aodh-dsvm-tempest-plugin-mysql/22bc757/ : SUCCESS in 41m 51s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/14/654314/1/check/aodh-dsvm-tempest-plugin-postgresql/4cf9b63/ : SUCCESS in 44m 41s\n- telemetry-dsvm-integration http://logs.openstack.org/14/654314/1/check/telemetry-dsvm-integration/79392d4/ : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":1},{"id":"af73f123a9696dbd9947983a193b74985eb4471a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 07:20:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94395df1c967a4a9701b46760a6d5f425807c1f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:25:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/14/654314/1/gate/openstack-tox-pep8/3ee87cc/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/14/654314/1/gate/openstack-tox-py27/14787bb/ : SUCCESS in 7m 37s\n- openstack-tox-py36 http://logs.openstack.org/14/654314/1/gate/openstack-tox-py36/32172a4/ : SUCCESS in 6m 51s\n- openstack-tox-py37 http://logs.openstack.org/14/654314/1/gate/openstack-tox-py37/c09e20a/ : SUCCESS in 7m 08s\n- openstack-tox-docs http://logs.openstack.org/14/654314/1/gate/openstack-tox-docs/b14c2f4/html/ : SUCCESS in 3m 58s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/14/654314/1/gate/aodh-dsvm-tempest-plugin-mysql/0347250/ : SUCCESS in 59m 20s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/14/654314/1/gate/aodh-dsvm-tempest-plugin-postgresql/8b2dc49/ : SUCCESS in 44m 14s\n- telemetry-dsvm-integration http://logs.openstack.org/14/654314/1/gate/telemetry-dsvm-integration/1c86381/ : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4a0ae4a3cc5cd360f1f336938eda43063fce88fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 08:25:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65eb29477bd7f17859e5ee1d56175c41d70d9fd1","revisions":{"65eb29477bd7f17859e5ee1d56175c41d70d9fd1":{"kind":"REWORK","_number":1,"created":"2019-04-22 06:21:16.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/14/654314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/14/654314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/14/654314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/14/654314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/14/654314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/14/654314/1"}}},"commit":{"parents":[{"commit":"80fd5c0c0557a7311c593a2c6d08bc021b37dbf9","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/80fd5c0c0557a7311c593a2c6d08bc021b37dbf9"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 06:20:45.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-22 06:20:45.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nThorough replacement of git.openstack.org URLs with their opendev.org\ncounterparts.\n\nChange-Id: I5adbc8490a6ab23d080e60b813febd1334f3f4e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/65eb29477bd7f17859e5ee1d56175c41d70d9fd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/65eb29477bd7f17859e5ee1d56175c41d70d9fd1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
