)]}'
{"id":"openstack%2Faodh~662559","triplet_id":"openstack%2Faodh~master~I170267ddbed31f63717e1c8fe119b2557afc812e","project":"openstack/aodh","branch":"master","topic":"bug/1831181","hashtags":[],"change_id":"I170267ddbed31f63717e1c8fe119b2557afc812e","subject":"Add user_domain_name to keystone client.","status":"MERGED","created":"2019-05-31 19:41:12.000000000","updated":"2019-06-03 14:22:25.000000000","submitted":"2019-06-01 07:14:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662559-1559373243401-816b8fbd","meta_rev_id":"68195edcaff6e4a34550a84eef670356b372c042","_number":662559,"virtual_id_number":662559,"owner":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-06-01 07:14:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-06-01 06:22:16.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-06-01 06:22:16.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-01 06:22:16.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-06-01 07:14:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cf3cac219e636abaf5c13e83fe79e820430c8d2c","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-05-31 19:41:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70711dc1840062d9cb92eb1ab6a41c743df6f040","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 20:44:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/662559/1/check/openstack-tox-pep8/9795c0f/ : SUCCESS in 3m 43s\n- openstack-tox-py27 http://logs.openstack.org/59/662559/1/check/openstack-tox-py27/8b40b01/ : SUCCESS in 7m 21s\n- openstack-tox-py36 http://logs.openstack.org/59/662559/1/check/openstack-tox-py36/28fa3de/ : SUCCESS in 5m 57s\n- openstack-tox-py37 http://logs.openstack.org/59/662559/1/check/openstack-tox-py37/d9bf7ee/ : SUCCESS in 8m 27s\n- openstack-tox-docs http://logs.openstack.org/59/662559/1/check/openstack-tox-docs/8ec0390/html/ : SUCCESS in 3m 38s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/59/662559/1/check/aodh-dsvm-tempest-plugin-mysql/f526338/ : SUCCESS in 1h 01m 12s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/59/662559/1/check/aodh-dsvm-tempest-plugin-postgresql/e84f5cd/ : SUCCESS in 42m 07s\n- telemetry-dsvm-integration http://logs.openstack.org/59/662559/1/check/telemetry-dsvm-integration/fc9eb37/ : SUCCESS in 48m 38s","accounts_in_message":[],"_revision_number":1},{"id":"12c797281741d0ca949ae7cad34fbfcec874d06a","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-06-01 06:22:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"8c7b6971a56c07304595e623be57dc864e660d7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 06:22:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"171fac8407800b80c4f90a07c9151aa9f2ed67c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 07:14:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/59/662559/1/gate/openstack-tox-pep8/96bee86/ : SUCCESS in 3m 32s\n- openstack-tox-py27 http://logs.openstack.org/59/662559/1/gate/openstack-tox-py27/83378b0/ : SUCCESS in 6m 30s\n- openstack-tox-py36 http://logs.openstack.org/59/662559/1/gate/openstack-tox-py36/5f91168/ : SUCCESS in 6m 01s\n- openstack-tox-py37 http://logs.openstack.org/59/662559/1/gate/openstack-tox-py37/7821095/ : SUCCESS in 7m 11s\n- openstack-tox-docs http://logs.openstack.org/59/662559/1/gate/openstack-tox-docs/4cc1cb2/html/ : SUCCESS in 3m 41s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/59/662559/1/gate/aodh-dsvm-tempest-plugin-mysql/ba29f90/ : SUCCESS in 42m 24s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/59/662559/1/gate/aodh-dsvm-tempest-plugin-postgresql/31becb7/ : SUCCESS in 45m 15s\n- telemetry-dsvm-integration http://logs.openstack.org/59/662559/1/gate/telemetry-dsvm-integration/6906bb3/ : SUCCESS in 46m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1f29627ee3efdd2c5dcd48f1bfb42980bb38a113","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-01 07:14:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"30551c21477e753b80e711f1da461852d3e03fac","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-06-03 14:22:25.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 1b1b8e20c7e4872e2872182757e561125859c9f0","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","revisions":{"83fb96463c6ba55e536c3c47fa04253ab1b53eff":{"kind":"REWORK","_number":1,"created":"2019-05-31 19:41:12.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/59/662559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/59/662559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/662559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/662559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/59/662559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/59/662559/1"}}},"commit":{"parents":[{"commit":"c6f07bf92d77b9ca847058c9d39d8de252f035d7","subject":"Merge \"Support Heat auto-healing notifier\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c6f07bf92d77b9ca847058c9d39d8de252f035d7"}]}],"author":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2019-05-31 19:37:07.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2019-05-31 19:40:56.000000000","tz":-240},"subject":"Add user_domain_name to keystone client.","message":"Add user_domain_name to keystone client.\n\nUse user_domain_name as a fallback if no user_domain_id is provided,\nthe default keystone password auth uses only user_domain_id, if not\nprovided aodh will fail getting a token, add the extra\nuser_domain_name parameter to the authentication.\n\nCloses-Bug: #1831181\n\nChange-Id: I170267ddbed31f63717e1c8fe119b2557afc812e\nSigned-off-by: Jorge Niedbalski \u003cjorge.niedbalski@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
