)]}'
{"id":"openstack%2Faodh~662766","triplet_id":"openstack%2Faodh~stable%2Fqueens~I170267ddbed31f63717e1c8fe119b2557afc812e","project":"openstack/aodh","branch":"stable/queens","topic":"bug/1831181-stable/stein-stable/queens","hashtags":[],"change_id":"I170267ddbed31f63717e1c8fe119b2557afc812e","subject":"Add user_domain_name to keystone client.","status":"MERGED","created":"2019-06-03 14:22:45.000000000","updated":"2019-06-04 02:29:56.000000000","submitted":"2019-06-04 02:29:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662766-1559615396106-24414518","meta_rev_id":"e06697eec1094b406662506eb5441f95ab8dd8ad","_number":662766,"virtual_id_number":662766,"owner":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-06-04 02:29:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-06-04 01:34:03.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-06-04 01:34:03.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-04 01:34:03.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-06-04 02:29:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34aee32d96e7c32b290f7ec879ddd43dd3cf6b1f","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-06-03 14:22:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"446a6957d0258b5a324fab1d5a0bd6188954baa3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:18:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/66/662766/1/check/openstack-tox-pep8/4c6b9eb/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/66/662766/1/check/openstack-tox-py27/4c8212e/ : SUCCESS in 7m 20s\n- openstack-tox-py35 http://logs.openstack.org/66/662766/1/check/openstack-tox-py35/7854f25/ : SUCCESS in 9m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/66/662766/1/check/build-openstack-sphinx-docs/c6890d4/html/ : SUCCESS in 4m 10s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/66/662766/1/check/aodh-dsvm-tempest-plugin-mysql/4d31d0d/ : SUCCESS in 34m 48s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/66/662766/1/check/aodh-dsvm-tempest-plugin-postgresql/85cf592/ : SUCCESS in 30m 32s\n- telemetry-dsvm-integration http://logs.openstack.org/66/662766/1/check/telemetry-dsvm-integration/16d4c91/ : SUCCESS in 47m 17s","accounts_in_message":[],"_revision_number":1},{"id":"31e2226d61e31ec16b09e4c15be9cc159cf893cd","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-06-04 01:34:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8cc6096f9c792baee14d603e920b2ae85e7ebb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 01:34:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ed31b8eea512af6ac7b5c4630e792116ed0e3c5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 02:29:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/66/662766/1/gate/openstack-tox-pep8/dbb3fcc/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/66/662766/1/gate/openstack-tox-py27/3452aa1/ : SUCCESS in 6m 53s\n- openstack-tox-py35 http://logs.openstack.org/66/662766/1/gate/openstack-tox-py35/e0b8d47/ : SUCCESS in 6m 43s\n- build-openstack-sphinx-docs http://logs.openstack.org/66/662766/1/gate/build-openstack-sphinx-docs/a76b142/html/ : SUCCESS in 4m 46s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/66/662766/1/gate/aodh-dsvm-tempest-plugin-mysql/8c74f90/ : SUCCESS in 27m 21s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/66/662766/1/gate/aodh-dsvm-tempest-plugin-postgresql/aae8cbd/ : SUCCESS in 37m 42s\n- telemetry-dsvm-integration http://logs.openstack.org/66/662766/1/gate/telemetry-dsvm-integration/e31df69/ : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3061d6f91e27c00bdc35918b6a244c462b56d274","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 02:29:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fce6e7b430c4732425e2ab41ec2bd231abda4840","revisions":{"fce6e7b430c4732425e2ab41ec2bd231abda4840":{"kind":"REWORK","_number":1,"created":"2019-06-03 14:22:45.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/66/662766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/66/662766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/662766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/662766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/66/662766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/66/662766/1"}}},"commit":{"parents":[{"commit":"f89a586584a19f48568ac1b6f729b63b0a8fe02b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f89a586584a19f48568ac1b6f729b63b0a8fe02b"}]}],"author":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2019-05-31 19:37:07.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2019-06-03 14:22:45.000000000","tz":0},"subject":"Add user_domain_name to keystone client.","message":"Add user_domain_name to keystone client.\n\nUse user_domain_name as a fallback if no user_domain_id is provided,\nthe default keystone password auth uses only user_domain_id, if not\nprovided aodh will fail getting a token, add the extra\nuser_domain_name parameter to the authentication.\n\nCloses-Bug: #1831181\n\nChange-Id: I170267ddbed31f63717e1c8fe119b2557afc812e\nSigned-off-by: Jorge Niedbalski \u003cjorge.niedbalski@canonical.com\u003e\n(cherry picked from commit 83fb96463c6ba55e536c3c47fa04253ab1b53eff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fce6e7b430c4732425e2ab41ec2bd231abda4840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/fce6e7b430c4732425e2ab41ec2bd231abda4840"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
