)]}'
{"id":"openstack%2Faodh~665022","triplet_id":"openstack%2Faodh~master~Idee3216c7c1b4f6391d08649637e619ba1a0ac13","project":"openstack/aodh","branch":"master","topic":"bug_resolve","hashtags":[],"change_id":"Idee3216c7c1b4f6391d08649637e619ba1a0ac13","subject":"Byte encoding of member Id","status":"MERGED","created":"2019-06-12 21:41:42.000000000","updated":"2019-06-28 05:23:28.000000000","submitted":"2019-06-28 05:23:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665022-1561699408482-31ea3c29","meta_rev_id":"1259ae60a66a4fc2299f75411db7a672579bc5bd","_number":665022,"virtual_id_number":665022,"owner":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-06-28 05:23:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},{"value":2,"date":"2019-06-28 00:51:08.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-06-28 04:25:21.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":1,"date":"2019-06-28 04:25:21.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 21:18:06.000000000","updated_by":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"reviewer":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"state":"REVIEWER"},{"updated":"2019-06-28 00:51:08.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-06-28 04:25:21.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-06-28 05:23:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a278d06b7fc104748e3d6c1ae0220cf7f56bd06b","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-06-12 21:41:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04ecda99b248a744135771d2ff8ebace0351fa43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 22:53:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/665022/1/check/openstack-tox-pep8/0e5448b/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/22/665022/1/check/openstack-tox-py27/bbf2d6e/ : SUCCESS in 7m 27s\n- openstack-tox-py36 http://logs.openstack.org/22/665022/1/check/openstack-tox-py36/17ac094/ : SUCCESS in 6m 41s\n- openstack-tox-py37 http://logs.openstack.org/22/665022/1/check/openstack-tox-py37/ffadb10/ : SUCCESS in 7m 34s\n- openstack-tox-docs http://logs.openstack.org/22/665022/1/check/openstack-tox-docs/623b3f9/html/ : SUCCESS in 4m 00s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/22/665022/1/check/aodh-dsvm-tempest-plugin-mysql/1daab47/ : SUCCESS in 1h 02m 36s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/22/665022/1/check/aodh-dsvm-tempest-plugin-postgresql/f7d2012/ : SUCCESS in 43m 59s\n- telemetry-dsvm-integration http://logs.openstack.org/22/665022/1/check/telemetry-dsvm-integration/2dbdae8/ : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4f68a7e2503b7acd2fbab6b2f1ad4820882a43aa","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2019-06-13 14:56:10.000000000","message":"Patch Set 1: Code-Review+2\n\ncan you open a bug with traceback you ran into for the record.","accounts_in_message":[],"_revision_number":1},{"id":"1ab1ac05d685d121fd11ba65d622a047e40747a6","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2019-06-13 14:56:51.000000000","message":"Patch Set 1:\n\nit would also be good to add test coverage for this so we can catch in future.","accounts_in_message":[],"_revision_number":1},{"id":"01b89ca3bbfd52278632545b276a42809d8d4bdb","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-06-14 18:05:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"09d9da6b8d7e3ad13ce6bdbbfa769cb12ff69984","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 19:05:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/22/665022/2/check/openstack-tox-pep8/e629e0b/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/22/665022/2/check/openstack-tox-py27/ca8dce2/ : SUCCESS in 6m 22s\n- openstack-tox-py36 http://logs.openstack.org/22/665022/2/check/openstack-tox-py36/b59e60e/ : FAILURE in 4m 21s\n- openstack-tox-py37 http://logs.openstack.org/22/665022/2/check/openstack-tox-py37/6b0dc7f/ : FAILURE in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/22/665022/2/check/openstack-tox-docs/22e917d/html/ : SUCCESS in 4m 10s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/22/665022/2/check/aodh-dsvm-tempest-plugin-mysql/1f25a2d/ : SUCCESS in 40m 38s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/22/665022/2/check/aodh-dsvm-tempest-plugin-postgresql/cd003a8/ : SUCCESS in 38m 36s\n- telemetry-dsvm-integration http://logs.openstack.org/22/665022/2/check/telemetry-dsvm-integration/933ef3a/ : SUCCESS in 47m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ec6f316bf4e69f551030c14c9cd01405ffaa1ddd","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-06-14 19:30:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1e4ed355987e3a567e704d906f167dbdc371f26e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 20:45:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/665022/3/check/openstack-tox-pep8/23be48d/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/22/665022/3/check/openstack-tox-py27/08e4ed8/ : SUCCESS in 7m 42s\n- openstack-tox-py36 http://logs.openstack.org/22/665022/3/check/openstack-tox-py36/96a5257/ : SUCCESS in 6m 46s\n- openstack-tox-py37 http://logs.openstack.org/22/665022/3/check/openstack-tox-py37/1dff81b/ : SUCCESS in 7m 06s\n- openstack-tox-docs http://logs.openstack.org/22/665022/3/check/openstack-tox-docs/84cc918/html/ : SUCCESS in 4m 02s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/22/665022/3/check/aodh-dsvm-tempest-plugin-mysql/c940fb9/ : SUCCESS in 44m 49s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/22/665022/3/check/aodh-dsvm-tempest-plugin-postgresql/3b6c2a8/ : SUCCESS in 49m 03s\n- telemetry-dsvm-integration http://logs.openstack.org/22/665022/3/check/telemetry-dsvm-integration/1ed661a/ : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":3},{"id":"cd64efbc8d0e90edb59bf96206c0a34718b822ab","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-06-15 00:54:41.000000000","message":"Patch Set 3:\n\nHi Vinay. Could you please add a test as Pradeep suggested? Thanks.","accounts_in_message":[],"_revision_number":3},{"id":"dab7b99885ec7fbe514a79baaf36c2d773bf073b","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-06-20 19:40:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5dac835b717e0f3044ece238e48e6162b7ffa23c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 20:41:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/665022/4/check/openstack-tox-pep8/53db7b4/ : SUCCESS in 3m 45s\n- openstack-tox-py27 http://logs.openstack.org/22/665022/4/check/openstack-tox-py27/6ff365d/ : SUCCESS in 9m 49s\n- openstack-tox-py36 http://logs.openstack.org/22/665022/4/check/openstack-tox-py36/04eaa37/ : SUCCESS in 10m 42s\n- openstack-tox-py37 http://logs.openstack.org/22/665022/4/check/openstack-tox-py37/83766ea/ : SUCCESS in 9m 33s\n- openstack-tox-docs http://logs.openstack.org/22/665022/4/check/openstack-tox-docs/d7eef4e/html/ : SUCCESS in 4m 31s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/22/665022/4/check/aodh-dsvm-tempest-plugin-mysql/7da2247/ : SUCCESS in 38m 42s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/22/665022/4/check/aodh-dsvm-tempest-plugin-postgresql/8928262/ : SUCCESS in 39m 11s\n- telemetry-dsvm-integration http://logs.openstack.org/22/665022/4/check/telemetry-dsvm-integration/3549a2b/ : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":4},{"id":"87b9d44916758cdc884d5fab0bfca40e21d3d75e","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2019-06-20 21:17:40.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e4c44b3c26854b94017eafec96a545cdf44af771","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2019-06-20 21:18:06.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"57ae35bdabe23ead1cc83d586e89b91faabfd115","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-06-20 21:28:25.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"ade2d8c0d6ce54006765a6a857d08f21ff742b42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:25:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/665022/5/check/openstack-tox-pep8/ecd0866/ : SUCCESS in 4m 01s\n- openstack-tox-py27 http://logs.openstack.org/22/665022/5/check/openstack-tox-py27/232dc79/ : SUCCESS in 6m 50s\n- openstack-tox-py36 http://logs.openstack.org/22/665022/5/check/openstack-tox-py36/33831e0/ : SUCCESS in 6m 34s\n- openstack-tox-py37 http://logs.openstack.org/22/665022/5/check/openstack-tox-py37/3a8e4a2/ : SUCCESS in 7m 18s\n- openstack-tox-docs http://logs.openstack.org/22/665022/5/check/openstack-tox-docs/c9ad7dc/html/ : SUCCESS in 4m 00s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/22/665022/5/check/aodh-dsvm-tempest-plugin-mysql/7e59d62/ : SUCCESS in 44m 41s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/22/665022/5/check/aodh-dsvm-tempest-plugin-postgresql/160899e/ : SUCCESS in 44m 46s\n- telemetry-dsvm-integration http://logs.openstack.org/22/665022/5/check/telemetry-dsvm-integration/e1ba049/ : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":5},{"id":"89987a784d77eddc160a0bee02c98556b76e41b9","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-24 02:32:08.000000000","message":"Patch Set 5: Code-Review+1\n\nCan we use the safe_encode from oslo_utils encodeutils to encode the value?","accounts_in_message":[],"_revision_number":5},{"id":"ae04f3ccb08ad610ccab1909a4485ad6f23d51a8","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-06-26 20:25:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1f676ac0f117e884e51d7ab48ad088edf07d1192","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 21:24:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/22/665022/6/check/openstack-tox-pep8/454e907/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/22/665022/6/check/openstack-tox-py27/c1ec473/ : SUCCESS in 6m 27s\n- openstack-tox-py36 http://logs.openstack.org/22/665022/6/check/openstack-tox-py36/27a557b/ : SUCCESS in 6m 04s\n- openstack-tox-py37 http://logs.openstack.org/22/665022/6/check/openstack-tox-py37/6dfa0cd/ : SUCCESS in 6m 49s\n- openstack-tox-docs http://logs.openstack.org/22/665022/6/check/openstack-tox-docs/34e43cf/html/ : SUCCESS in 4m 22s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/22/665022/6/check/aodh-dsvm-tempest-plugin-mysql/f3cfee8/ : SUCCESS in 39m 04s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/22/665022/6/check/aodh-dsvm-tempest-plugin-postgresql/1a7984b/ : SUCCESS in 41m 54s\n- telemetry-dsvm-integration http://logs.openstack.org/22/665022/6/check/telemetry-dsvm-integration/4818bdd/ : FAILURE in 52m 47s","accounts_in_message":[],"_revision_number":6},{"id":"7a6cb9ed0298d1687fd2863eee0a5ec99c800207","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-06-26 22:00:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"47d61facfb2964433ff2d7fa740004b85c68b6b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:59:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/665022/7/check/openstack-tox-pep8/dc2360e/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/22/665022/7/check/openstack-tox-py27/9f001ec/ : SUCCESS in 8m 34s\n- openstack-tox-py36 http://logs.openstack.org/22/665022/7/check/openstack-tox-py36/a80a8b2/ : SUCCESS in 6m 50s\n- openstack-tox-py37 http://logs.openstack.org/22/665022/7/check/openstack-tox-py37/4a14bea/ : SUCCESS in 6m 29s\n- openstack-tox-docs http://logs.openstack.org/22/665022/7/check/openstack-tox-docs/5f4cac1/html/ : SUCCESS in 3m 59s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/22/665022/7/check/aodh-dsvm-tempest-plugin-mysql/9dea95d/ : SUCCESS in 41m 26s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/22/665022/7/check/aodh-dsvm-tempest-plugin-postgresql/e7d9058/ : SUCCESS in 41m 49s\n- telemetry-dsvm-integration http://logs.openstack.org/22/665022/7/check/telemetry-dsvm-integration/2447794/ : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":7},{"id":"112432b1b6c919932f6f096c236cf2f0d346f648","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-27 03:39:00.000000000","message":"Patch Set 7: Code-Review-1\n\nI cherry picked you PS6 with safe_encode, all tests are passed. shall we changed to use safe_encode?\n\nhttps://review.opendev.org/#/c/667771/","accounts_in_message":[],"_revision_number":7},{"id":"733b85add8b87e8d4a975fd805303cfd8c895cab","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-06-27 19:28:31.000000000","message":"Patch Set 7:\n\n\u003e I cherry picked you PS6 with safe_encode, all tests are passed.\n \u003e shall we changed to use safe_encode?\n \u003e \n \u003e https://review.opendev.org/#/c/667771/\n\nSure, I\u0027ll add myself and pradeep to review the patch.","accounts_in_message":[],"_revision_number":7},{"id":"322fcbdfae34bf4ff98631246fda1beebe98c8bd","author":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"date":"2019-06-27 19:43:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1ea9fcd1ea985aa0688ea6613ec0b2b79ffe48a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 20:43:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/665022/8/check/openstack-tox-pep8/7ddbf43/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/22/665022/8/check/openstack-tox-py27/17c8e23/ : SUCCESS in 6m 12s\n- openstack-tox-py36 http://logs.openstack.org/22/665022/8/check/openstack-tox-py36/99c52f4/ : SUCCESS in 7m 03s\n- openstack-tox-py37 http://logs.openstack.org/22/665022/8/check/openstack-tox-py37/42fa6e1/ : SUCCESS in 7m 26s\n- openstack-tox-docs http://logs.openstack.org/22/665022/8/check/openstack-tox-docs/cb7f90b/html/ : SUCCESS in 3m 29s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/22/665022/8/check/aodh-dsvm-tempest-plugin-mysql/45adde9/ : SUCCESS in 37m 38s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/22/665022/8/check/aodh-dsvm-tempest-plugin-postgresql/e3a2f6b/ : SUCCESS in 38m 53s\n- telemetry-dsvm-integration http://logs.openstack.org/22/665022/8/check/telemetry-dsvm-integration/05b7e4c/ : SUCCESS in 48m 28s","accounts_in_message":[],"_revision_number":8},{"id":"2d9ef13a566e0a49bde9f6ff052b1fb22ccd3a8c","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-06-28 00:51:08.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"02d3f92f0ad968c5afa3de24fcb2bb7cd2adefab","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-06-28 04:25:21.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":8},{"id":"6d6df5102b498271fb48dfee6e428c3b8c96f86f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 04:25:34.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f394169d6ec8e06b384cbd1f0f3f5dbaa3e83ccf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 05:23:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"bdc227d27cb51ab8f0688255f821ce220050052c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 05:23:28.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/665022/8/gate/openstack-tox-pep8/0ceb356/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/22/665022/8/gate/openstack-tox-py27/a3ad244/ : SUCCESS in 8m 10s\n- openstack-tox-py36 http://logs.openstack.org/22/665022/8/gate/openstack-tox-py36/e02c675/ : SUCCESS in 6m 03s\n- openstack-tox-py37 http://logs.openstack.org/22/665022/8/gate/openstack-tox-py37/c49fab4/ : SUCCESS in 8m 14s\n- openstack-tox-docs http://logs.openstack.org/22/665022/8/gate/openstack-tox-docs/2ef8ee2/html/ : SUCCESS in 4m 31s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/22/665022/8/gate/aodh-dsvm-tempest-plugin-mysql/637bacb/ : SUCCESS in 39m 35s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/22/665022/8/gate/aodh-dsvm-tempest-plugin-postgresql/d7ea0d9/ : SUCCESS in 39m 31s\n- telemetry-dsvm-integration http://logs.openstack.org/22/665022/8/gate/telemetry-dsvm-integration/f4a5440/ : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"9c9112ad1f07fad457cb76612a43394ece840e06","revisions":{"c470c3e186847a08bc448e0250b57c3e6b650ca4":{"kind":"REWORK","_number":1,"created":"2019-06-12 21:41:42.000000000","uploader":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"ref":"refs/changes/22/665022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/22/665022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/22/665022/1"}}},"commit":{"parents":[{"commit":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]}],"author":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-12 21:36:15.000000000","tz":-240},"committer":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-12 21:36:15.000000000","tz":-240},"subject":"Ascii encoding of Id","message":"Ascii encoding of Id\n\nWhen calling for class PartitionCoordinator\nit expects the member id to be encoded in\nascii otherwise it raises MemberNotInGroupError\nexception.\n\nChange-Id: Idee3216c7c1b4f6391d08649637e619ba1a0ac13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c470c3e186847a08bc448e0250b57c3e6b650ca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c470c3e186847a08bc448e0250b57c3e6b650ca4"}]},"branch":"refs/heads/master"},"2d931298c2aad74687a8e84f3bd41567477469e9":{"kind":"REWORK","_number":2,"created":"2019-06-14 18:05:29.000000000","uploader":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"ref":"refs/changes/22/665022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/22/665022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/22/665022/2"}}},"commit":{"parents":[{"commit":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]}],"author":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-12 21:36:15.000000000","tz":-240},"committer":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-14 18:05:10.000000000","tz":-240},"subject":"Byte encoding of member Id","message":"Byte encoding of member Id\n\nFrom OSP15 we will be porting to\npython3 due to rhel8, to deal with that change,\nwhen calling for class PartitionCoordinator\nthe member id need to be encoded in\nbyte format otherwise it raises MemberNotInGroupError\nexception.\n\nCloses-Bug: https://storyboard.openstack.org/#!/story/2005894\nChange-Id: Idee3216c7c1b4f6391d08649637e619ba1a0ac13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2d931298c2aad74687a8e84f3bd41567477469e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2d931298c2aad74687a8e84f3bd41567477469e9"}]},"branch":"refs/heads/master"},"22fe7526a1282fbaeaedc9c8a6362bcbf9d8904c":{"kind":"REWORK","_number":3,"created":"2019-06-14 19:30:52.000000000","uploader":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"ref":"refs/changes/22/665022/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/22/665022/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/22/665022/3"}}},"commit":{"parents":[{"commit":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]}],"author":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-12 21:36:15.000000000","tz":-240},"committer":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-14 19:30:03.000000000","tz":-240},"subject":"Byte encoding of member Id","message":"Byte encoding of member Id\n\nFrom OSP15 we will be porting to\npython3 due to rhel8, to deal with that change,\nwhen calling for class PartitionCoordinator\nthe member id need to be encoded in\nbyte format otherwise it raises MemberNotInGroupError\nexception.\n\nCloses-Bug: https://storyboard.openstack.org/#!/story/2005894\nChange-Id: Idee3216c7c1b4f6391d08649637e619ba1a0ac13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/22fe7526a1282fbaeaedc9c8a6362bcbf9d8904c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/22fe7526a1282fbaeaedc9c8a6362bcbf9d8904c"}]},"branch":"refs/heads/master"},"374f4c5d6fa025570cc6707b18fe0b07d29f5ab4":{"kind":"REWORK","_number":4,"created":"2019-06-20 19:40:51.000000000","uploader":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"ref":"refs/changes/22/665022/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/22/665022/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/22/665022/4"}}},"commit":{"parents":[{"commit":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]}],"author":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-12 21:36:15.000000000","tz":-240},"committer":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-20 19:40:16.000000000","tz":-240},"subject":"Byte encoding of member Id","message":"Byte encoding of member Id\n\nFrom OSP15 we will be porting to\npython3 due to rhel8, to deal with that change,\nwhen calling for class PartitionCoordinator\nthe member id need to be encoded in\nbyte format otherwise it raises MemberNotInGroupError\nexception.\n\nCloses-Bug: https://storyboard.openstack.org/#!/story/2005894\nChange-Id: Idee3216c7c1b4f6391d08649637e619ba1a0ac13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/374f4c5d6fa025570cc6707b18fe0b07d29f5ab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/374f4c5d6fa025570cc6707b18fe0b07d29f5ab4"}]},"branch":"refs/heads/master"},"f359a2b21d6317e97cffc41a10bec33a54c9731e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-06-20 21:28:25.000000000","uploader":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"ref":"refs/changes/22/665022/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/22/665022/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/22/665022/5"}}},"commit":{"parents":[{"commit":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]}],"author":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-12 21:36:15.000000000","tz":-240},"committer":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-20 21:22:52.000000000","tz":-240},"subject":"Byte encoding of member Id","message":"Byte encoding of member Id\n\nFrom stein we will be porting to python3.\nTo deal with that, when calling for class PartitionCoordinator\nthe member id need to be encoded in byte format\notherwise it raises MemberNotInGroupError exception.\n\nCloses-Bug: https://storyboard.openstack.org/#!/story/2005894\nChange-Id: Idee3216c7c1b4f6391d08649637e619ba1a0ac13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f359a2b21d6317e97cffc41a10bec33a54c9731e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f359a2b21d6317e97cffc41a10bec33a54c9731e"}]},"branch":"refs/heads/master"},"bf2cf7d86181a2f7e975a07efa31b48323a12a98":{"kind":"REWORK","_number":6,"created":"2019-06-26 20:25:22.000000000","uploader":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"ref":"refs/changes/22/665022/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/22/665022/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/22/665022/6"}}},"commit":{"parents":[{"commit":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]}],"author":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-12 21:36:15.000000000","tz":-240},"committer":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-26 20:24:43.000000000","tz":-240},"subject":"Byte encoding of member Id","message":"Byte encoding of member Id\n\nFrom stein we will be porting to python3.\nTo deal with that, when calling for class PartitionCoordinator\nthe member id need to be encoded in byte format\notherwise it raises MemberNotInGroupError exception.\n\nCloses-Bug: https://storyboard.openstack.org/#!/story/2005894\nChange-Id: Idee3216c7c1b4f6391d08649637e619ba1a0ac13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bf2cf7d86181a2f7e975a07efa31b48323a12a98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bf2cf7d86181a2f7e975a07efa31b48323a12a98"}]},"branch":"refs/heads/master"},"f21c0b5cf7ff073afcf1164da0e9b4eb8c2f76e2":{"kind":"REWORK","_number":7,"created":"2019-06-26 22:00:19.000000000","uploader":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"ref":"refs/changes/22/665022/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/22/665022/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/22/665022/7"}}},"commit":{"parents":[{"commit":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]}],"author":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-12 21:36:15.000000000","tz":-240},"committer":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-26 22:00:12.000000000","tz":-240},"subject":"Byte encoding of member Id","message":"Byte encoding of member Id\n\nFrom stein we will be porting to python3.\nTo deal with that, when calling for class PartitionCoordinator\nthe member id need to be encoded in byte format\notherwise it raises MemberNotInGroupError exception.\n\nCloses-Bug: https://storyboard.openstack.org/#!/story/2005894\nChange-Id: Idee3216c7c1b4f6391d08649637e619ba1a0ac13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f21c0b5cf7ff073afcf1164da0e9b4eb8c2f76e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f21c0b5cf7ff073afcf1164da0e9b4eb8c2f76e2"}]},"branch":"refs/heads/master"},"9c9112ad1f07fad457cb76612a43394ece840e06":{"kind":"REWORK","_number":8,"created":"2019-06-27 19:43:48.000000000","uploader":{"_account_id":28504,"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","username":"vkapalav"},"ref":"refs/changes/22/665022/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/22/665022/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/22/665022/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/22/665022/8"}}},"commit":{"parents":[{"commit":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]}],"author":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-12 21:36:15.000000000","tz":-240},"committer":{"name":"Vinay Kapalavai","email":"vkapalav@redhat.com","date":"2019-06-27 19:43:47.000000000","tz":-240},"subject":"Byte encoding of member Id","message":"Byte encoding of member Id\n\nFrom stein we will be porting to python3.\nTo deal with that, when calling for class PartitionCoordinator\nthe member id need to be encoded in byte format\notherwise it raises MemberNotInGroupError exception.\n\nCloses-Bug: https://storyboard.openstack.org/#!/story/2005894\nChange-Id: Idee3216c7c1b4f6391d08649637e619ba1a0ac13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9c9112ad1f07fad457cb76612a43394ece840e06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9c9112ad1f07fad457cb76612a43394ece840e06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
