)]}'
{"id":"openstack%2Faodh~665991","triplet_id":"openstack%2Faodh~master~I24cb8d8d084a291028a961823fcac4f09b4bf1ca","project":"openstack/aodh","branch":"master","hashtags":[],"change_id":"I24cb8d8d084a291028a961823fcac4f09b4bf1ca","subject":"Sync Sphinx requirement","status":"MERGED","created":"2019-06-18 14:10:11.000000000","updated":"2019-06-20 02:42:09.000000000","submitted":"2019-06-19 08:50:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665991-1560934215909-fc442d7a","meta_rev_id":"ccd0215e69ab6dddbab4e7009420de011522aff0","_number":665991,"virtual_id_number":665991,"owner":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-06-19 08:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-06-19 07:59:43.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-06-19 07:59:43.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-19 07:59:43.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-06-19 08:50:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4905e95899f072eeaf1faa234a268bcf7b24906a","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-18 14:10:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39925593636a8029ada5e0709b006d0d6c0f4800","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 15:05:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/91/665991/1/check/test-release-openstack/d5c3c46/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/91/665991/1/check/openstack-tox-pep8/364f277/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/91/665991/1/check/openstack-tox-py27/53693da/ : SUCCESS in 10m 37s\n- openstack-tox-py36 http://logs.openstack.org/91/665991/1/check/openstack-tox-py36/7772e80/ : FAILURE in 7m 35s\n- openstack-tox-py37 http://logs.openstack.org/91/665991/1/check/openstack-tox-py37/9773339/ : SUCCESS in 10m 02s\n- openstack-tox-docs http://logs.openstack.org/91/665991/1/check/openstack-tox-docs/835c099/html/ : SUCCESS in 4m 29s\n- build-openstack-releasenotes http://logs.openstack.org/91/665991/1/check/build-openstack-releasenotes/48ea68a/html/ : SUCCESS in 8m 08s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/91/665991/1/check/aodh-dsvm-tempest-plugin-mysql/d4c233b/ : SUCCESS in 41m 15s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/91/665991/1/check/aodh-dsvm-tempest-plugin-postgresql/8967a87/ : SUCCESS in 41m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3ba4b115fb2ef819c3f692181d9b22b53a8e6350","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-06-19 01:51:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nWe\u0027ve moved to Storyboard. Please update the link as suggested. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"b15c889ad37bd55346ea5f5d3582c458c1be0f4c","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-19 06:46:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5807b5c13219e70f6fba0fbc7671576f39cdc64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 07:55:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/91/665991/2/check/test-release-openstack/aa3fc26/ : SUCCESS in 2m 41s\n- openstack-tox-pep8 http://logs.openstack.org/91/665991/2/check/openstack-tox-pep8/b81c8a2/ : SUCCESS in 3m 56s\n- openstack-tox-py27 http://logs.openstack.org/91/665991/2/check/openstack-tox-py27/3629ac3/ : SUCCESS in 6m 18s\n- openstack-tox-py36 http://logs.openstack.org/91/665991/2/check/openstack-tox-py36/fbd5947/ : SUCCESS in 6m 49s\n- openstack-tox-py37 http://logs.openstack.org/91/665991/2/check/openstack-tox-py37/3db792d/ : SUCCESS in 7m 21s\n- openstack-tox-docs http://logs.openstack.org/91/665991/2/check/openstack-tox-docs/44e9fe1/html/ : SUCCESS in 4m 24s\n- build-openstack-releasenotes http://logs.openstack.org/91/665991/2/check/build-openstack-releasenotes/62ead7c/html/ : SUCCESS in 8m 00s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/91/665991/2/check/aodh-dsvm-tempest-plugin-mysql/4ffe169/ : SUCCESS in 49m 11s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/91/665991/2/check/aodh-dsvm-tempest-plugin-postgresql/c0c60bd/ : SUCCESS in 41m 33s","accounts_in_message":[],"_revision_number":2},{"id":"febe7e01060031e45ee78cfab46b7b5dab81bbf2","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-06-19 07:59:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"5b7b0a78442efe7024478f202a74aa8cce7e0e8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 07:59:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a4f9fb712678fa354bbf037622ec8476b255f373","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 08:50:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ac72b12df50f2e094bcf5f6dfb72ba8439899fee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 08:50:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/91/665991/2/gate/test-release-openstack/c5bb8ff/ : SUCCESS in 2m 40s\n- openstack-tox-pep8 http://logs.openstack.org/91/665991/2/gate/openstack-tox-pep8/939ed72/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/91/665991/2/gate/openstack-tox-py27/d17d903/ : SUCCESS in 6m 44s\n- openstack-tox-py36 http://logs.openstack.org/91/665991/2/gate/openstack-tox-py36/cd106d7/ : SUCCESS in 9m 14s\n- openstack-tox-py37 http://logs.openstack.org/91/665991/2/gate/openstack-tox-py37/dcbc26a/ : SUCCESS in 10m 27s\n- openstack-tox-docs http://logs.openstack.org/91/665991/2/gate/openstack-tox-docs/e51be2b/html/ : SUCCESS in 3m 39s\n- build-openstack-releasenotes http://logs.openstack.org/91/665991/2/gate/build-openstack-releasenotes/9b72709/html/ : SUCCESS in 8m 09s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/91/665991/2/gate/aodh-dsvm-tempest-plugin-mysql/390768a/ : SUCCESS in 40m 11s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/91/665991/2/gate/aodh-dsvm-tempest-plugin-postgresql/0638ef8/ : SUCCESS in 43m 50s","accounts_in_message":[],"_revision_number":2},{"id":"ccd0215e69ab6dddbab4e7009420de011522aff0","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-20 02:42:09.000000000","message":"Patch Set 2:\n\n\u003e Change has been successfully merged by Zuul\n\nthanks for review, and please help review https://review.opendev.org/#/c/666248/","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f3ffbb4dce38b0c7430dfd6c02fc28760557a63d","revisions":{"8734b3a998d16a752c60212ca3885d1ec7dddf16":{"kind":"REWORK","_number":1,"created":"2019-06-18 14:10:11.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/91/665991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/91/665991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/665991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/665991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/665991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/91/665991/1"}}},"commit":{"parents":[{"commit":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-18 14:09:35.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-18 14:10:07.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Update some URLs to latest\n\nChange-Id: I24cb8d8d084a291028a961823fcac4f09b4bf1ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8734b3a998d16a752c60212ca3885d1ec7dddf16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8734b3a998d16a752c60212ca3885d1ec7dddf16"}]},"branch":"refs/heads/master"},"f3ffbb4dce38b0c7430dfd6c02fc28760557a63d":{"kind":"REWORK","_number":2,"created":"2019-06-19 06:46:47.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/91/665991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/91/665991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/665991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/665991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/665991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/91/665991/2"}}},"commit":{"parents":[{"commit":"83fb96463c6ba55e536c3c47fa04253ab1b53eff","subject":"Add user_domain_name to keystone client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/83fb96463c6ba55e536c3c47fa04253ab1b53eff"}]}],"author":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-18 14:09:35.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2019-06-19 06:46:28.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\n1. Sync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n2. Update some URLs to latest\n\nChange-Id: I24cb8d8d084a291028a961823fcac4f09b4bf1ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f3ffbb4dce38b0c7430dfd6c02fc28760557a63d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f3ffbb4dce38b0c7430dfd6c02fc28760557a63d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
