)]}'
{"id":"openstack%2Faodh~669515","triplet_id":"openstack%2Faodh~master~Ife7bbcb1f5d4f766ebf9470e252a8c0cf520ee16","project":"openstack/aodh","branch":"master","topic":"python3-train","hashtags":[],"change_id":"Ife7bbcb1f5d4f766ebf9470e252a8c0cf520ee16","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 20:27:33.000000000","updated":"2019-07-06 04:14:49.000000000","submitted":"2019-07-06 04:14:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669515-1562386489921-6efaa91b","meta_rev_id":"95b5d7f7e2082f1f11b1da98f31d70706fec6230","_number":669515,"virtual_id_number":669515,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-07-06 04:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":2,"date":"2019-07-06 03:17:41.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":1,"date":"2019-07-06 03:17:41.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-06 03:17:41.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-07-06 04:14:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"840a08bbd1818405850d19d76b6b9c1479507422","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 20:27:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b721a353d2c103189c46c6027efdc6d819155de","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 20:29:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"59b60aa231ed3d7f612eae15fdd599041d766ce0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:26:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/15/669515/2/check/test-release-openstack/3186eb6/ : SUCCESS in 2m 39s\n- openstack-tox-pep8 http://logs.openstack.org/15/669515/2/check/openstack-tox-pep8/2ae435d/ : SUCCESS in 3m 51s\n- openstack-tox-py27 http://logs.openstack.org/15/669515/2/check/openstack-tox-py27/6420f10/ : SUCCESS in 7m 21s\n- openstack-tox-py36 http://logs.openstack.org/15/669515/2/check/openstack-tox-py36/e913b3b/ : SUCCESS in 7m 07s\n- openstack-tox-py37 http://logs.openstack.org/15/669515/2/check/openstack-tox-py37/5d07500/ : SUCCESS in 6m 23s\n- openstack-tox-docs http://logs.openstack.org/15/669515/2/check/openstack-tox-docs/b5743c9/html/ : SUCCESS in 3m 37s\n- build-openstack-releasenotes http://logs.openstack.org/15/669515/2/check/build-openstack-releasenotes/644b7e7/html/ : SUCCESS in 8m 16s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/15/669515/2/check/aodh-dsvm-tempest-plugin-mysql/105dd57/ : SUCCESS in 46m 22s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/15/669515/2/check/aodh-dsvm-tempest-plugin-postgresql/e34cc51/ : SUCCESS in 39m 37s\n- telemetry-dsvm-integration http://logs.openstack.org/15/669515/2/check/telemetry-dsvm-integration/34e7867/ : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":2},{"id":"0cfd9b731d80ec3ec62b73fee24814a8c9ee4221","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-07-06 03:17:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks.","accounts_in_message":[],"_revision_number":2},{"id":"c4a4f1a839afca2485fb4e3fc97eadb8b485cf36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 03:17:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5837d9b4b72a86ec3436bd544fd72aede03ba8c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 04:14:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/15/669515/2/gate/test-release-openstack/f23ccec/ : SUCCESS in 2m 25s\n- openstack-tox-pep8 http://logs.openstack.org/15/669515/2/gate/openstack-tox-pep8/bea6d43/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/15/669515/2/gate/openstack-tox-py27/ede8824/ : SUCCESS in 8m 10s\n- openstack-tox-py36 http://logs.openstack.org/15/669515/2/gate/openstack-tox-py36/526d27d/ : SUCCESS in 6m 10s\n- openstack-tox-py37 http://logs.openstack.org/15/669515/2/gate/openstack-tox-py37/e4dcff6/ : SUCCESS in 7m 48s\n- openstack-tox-docs http://logs.openstack.org/15/669515/2/gate/openstack-tox-docs/d7d68b1/html/ : SUCCESS in 3m 34s\n- build-openstack-releasenotes http://logs.openstack.org/15/669515/2/gate/build-openstack-releasenotes/80784fa/html/ : SUCCESS in 6m 51s\n- aodh-dsvm-tempest-plugin-mysql http://logs.openstack.org/15/669515/2/gate/aodh-dsvm-tempest-plugin-mysql/ffa2523/ : SUCCESS in 40m 06s\n- aodh-dsvm-tempest-plugin-postgresql http://logs.openstack.org/15/669515/2/gate/aodh-dsvm-tempest-plugin-postgresql/f9696c2/ : SUCCESS in 40m 01s\n- telemetry-dsvm-integration http://logs.openstack.org/15/669515/2/gate/telemetry-dsvm-integration/5a7aba6/ : SUCCESS in 53m 51s","accounts_in_message":[],"_revision_number":2},{"id":"9ca0bad10d39f517e9c53aaad8c761a197ba137c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-06 04:14:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c45127a5c197edb98c5f0148fa77f7b91633127e","revisions":{"ccfdf2bdbdace613cd4ccb9213f7e1d3c399485a":{"kind":"REWORK","_number":1,"created":"2019-07-05 20:27:33.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/15/669515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/15/669515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/669515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/669515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/669515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/15/669515/1"}}},"commit":{"parents":[{"commit":"f64d58a34892255bac3e92821670e75f2ba3569f","subject":"Merge \"Byte encoding of member Id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f64d58a34892255bac3e92821670e75f2ba3569f"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:03:30.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 20:27:33.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ife7bbcb1f5d4f766ebf9470e252a8c0cf520ee16\nStory: #2005924\nTask: #34251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ccfdf2bdbdace613cd4ccb9213f7e1d3c399485a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ccfdf2bdbdace613cd4ccb9213f7e1d3c399485a"}]},"branch":"refs/heads/master"},"c45127a5c197edb98c5f0148fa77f7b91633127e":{"kind":"REWORK","_number":2,"created":"2019-07-05 20:29:55.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/15/669515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/15/669515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/669515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/669515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/669515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/15/669515/2"}}},"commit":{"parents":[{"commit":"f64d58a34892255bac3e92821670e75f2ba3569f","subject":"Merge \"Byte encoding of member Id\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f64d58a34892255bac3e92821670e75f2ba3569f"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 16:03:30.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 20:29:29.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: Ife7bbcb1f5d4f766ebf9470e252a8c0cf520ee16\nStory: #2005924\nTask: #34251\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c45127a5c197edb98c5f0148fa77f7b91633127e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c45127a5c197edb98c5f0148fa77f7b91633127e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
