)]}'
{"id":"openstack%2Faodh~678762","triplet_id":"openstack%2Faodh~master~I47d1753f2700c1339171c471a682718a5b72c9d6","project":"openstack/aodh","branch":"master","hashtags":[],"change_id":"I47d1753f2700c1339171c471a682718a5b72c9d6","subject":"Add the native zuulv3 jobs instead of legacy way","status":"MERGED","created":"2019-08-27 09:17:16.000000000","updated":"2019-09-03 05:25:12.000000000","submitted":"2019-09-03 05:22:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678762-1567488168795-1d226e13","meta_rev_id":"021e42e36223dec28c60f5eaf9382f25625ef31c","_number":678762,"virtual_id_number":678762,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-02 01:13:04.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":2,"date":"2019-09-03 05:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2019-09-03 04:18:09.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":1,"date":"2019-09-03 04:18:09.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-03 04:18:09.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-09-03 05:22:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03313cf71475be4b195aa102cbc91e4bece3becd","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-27 09:17:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e96117f2f20096ac08db78a77a16eb62b3efe872","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 10:23:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56a47585cd734a3fa004b07d87b576fe : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/700657b22a2a40ab9ebf5c879e23ed30 : SUCCESS in 10m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5a22134719441a3b0c5ed3075979921 : SUCCESS in 12m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bcf4e6d0de0d4c03a81467d9bb82dbf7 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9211c17984e4f66bcbb4351f6b4dc51 : FAILURE in 4m 49s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ee82aad661164e43a4dd31d0c12e4342 : SUCCESS in 48m 07s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/6f40b3d5e576496ead577a7a52283c0c : SUCCESS in 49m 06s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c86c74adc65342bb8fbb23fd1fe0f538 : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ab63bcad2f69b1533fc5ca254a6b08fe39c7ca81","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-27 11:01:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"707b9fe77b60cbc63e50a5bee88f23e3688b88dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 12:03:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aaec80f6ee944058e31409eeed637e2 : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f22bfd1cb1564aaa9832cdaf65ed71a2 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76e0cdbc7da64d9dbf701aaf260f4f04 : SUCCESS in 8m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/532f69692d4f44b2b79906caf9ec4bb8 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e14701b9edc4c758093be62fc25f71e : FAILURE in 5m 31s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f43067fad320496db0cd2695b25e0456 : SUCCESS in 53m 58s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/71220480840c479184b2db9a302005a4 : SUCCESS in 50m 07s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cb6abfb2dcd24fb68d9c39709344d482 : SUCCESS in 53m 45s","accounts_in_message":[],"_revision_number":2},{"id":"68a76686a680fa1682b5c12f25e6a98a70f720da","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-28 00:59:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d6dd9ec73c19dde57348076560f9d41b874c74de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 01:58:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd8e6a0d56d64fc8a5125d15c433880c : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70f79425914c4c0c9de3b7fb88d1c90d : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43fd677534f842ea99b022db12d3f7d0 : SUCCESS in 8m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24a5fc144ef34ff89c54affc621c98dc : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e020bb9a3d504d9e8c2d7c7887fecdd3 : FAILURE in 4m 11s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/771a16dc57354cb3a8e5a5d1809d59ea : SUCCESS in 46m 55s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/dd04938362c8460a89e4e66d6fff348c : FAILURE in 29m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3debc8cfd70249f8b34405038c48520a : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d258ac61dad9318f5b6ad3f4f7d8a6e3a8be1226","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-30 05:18:14.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"43ad5d77d5ff88b53905582b5e756f185c02b008","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 06:20:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c287d69735db4b15bffbb5b3fb1ef642 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8edde69ca1ee468a977a2e15b8d65e79 : SUCCESS in 8m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/352415a2762f4bb2a92b6c080d54a0c5 : SUCCESS in 9m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a896e6c384074fce91fce7a06d116cb1 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9908d183924a608932b25229fdd6d6 : SUCCESS in 6m 10s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/890fd0d99d3f4e9aa2a8592ce988be9d : SUCCESS in 52m 21s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/edf7100fcffd41648def34b0da0d5a4a : SUCCESS in 59m 03s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4e0f7d02ec5446a19cf0cbb889605894 : FAILURE in 55m 28s","accounts_in_message":[],"_revision_number":3},{"id":"17e1e5dc5ee69183dc533fe74dff01d325f42653","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-08-30 07:15:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"529d8a3203bf62de5835597851baeebca10fada0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 08:15:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03746420d1de4142b39691632a7c12ed : SUCCESS in 6m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73e9e3f665f64841bf0f487dfdb8f4a1 : FAILURE in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48b2d484e63143c89bb5703391c65440 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9fdfa1bfe1849b3b068878f2d55f92a : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a3c63025c14d7ea5337431b34bae3d : SUCCESS in 5m 26s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/bcb06ec6a1c24f8384db29e3bfa11a6b : SUCCESS in 51m 11s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/c01e59b54ef243db8da6b970c54bdb97 : SUCCESS in 48m 33s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4d7e695dd0f34a9fb35526d5b965a4d0 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":3},{"id":"86060563f50dde327a8a8ed022ae995d75f592d3","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-09-02 01:13:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b3af86453cf460583b50045043afd4e7e3e0f0bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 02:15:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e740061891a495cbe8c68dc14645e09 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0dd33fe36fc44f4c9c90f41478e7a20a : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/401817b25c9841f895fe94a34752a3b1 : SUCCESS in 9m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/110d3efb42404645a3f063969c7bc754 : SUCCESS in 8m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5ab9169c12c4fc893516a425fd21d85 : SUCCESS in 5m 46s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/498868edbea0421a80b3e59c51039f82 : SUCCESS in 56m 24s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b772a95df17440189528b9e829536ff5 : SUCCESS in 1h 00m 17s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/860a39775e424dc1b0bff2167cc7cb83 : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":3},{"id":"a6997e9c41191ed36127ae4624a9625f681cedc5","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2019-09-03 04:18:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":3},{"id":"e7fc2acaf2656ce1c8ecfe2b27e21063152356fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 04:18:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1b9bf93ae88cb7ebd95336268a1d56e0f7c446df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 05:22:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b124e020d2049dfa8566873e1a8f79a : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5feeb4c5268940b289abb19e10052899 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6847cac301bd4db9a22ae13bdac0b224 : SUCCESS in 7m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/02f5da8d57684241a093c791699d6ee3 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dbe69e6875842d88b25ad035328d692 : SUCCESS in 5m 02s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3c8b5b9f12654f02ae331dbe58fd6bc3 : SUCCESS in 48m 36s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/59105ba72eda47dcaa36ad3d2c418974 : SUCCESS in 49m 25s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f6cb733cef2545b49ede51cb6b98bb3e : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":3},{"id":"562a4a39a8dd40947ff388076eed1b778d99a065","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 05:22:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"021e42e36223dec28c60f5eaf9382f25625ef31c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 05:25:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a216f3295ef4f19abf5c20d6ad38001 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bc64b28a04ebdf7070e7297f7bce13e08425af31","revisions":{"402b5c77a9ad2e1221d96eb73ffaec79d2e50c2b":{"kind":"REWORK","_number":1,"created":"2019-08-27 09:17:16.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/62/678762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/62/678762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/678762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/678762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/678762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/62/678762/1"}}},"commit":{"parents":[{"commit":"e2d2ebf44ebb0e5120566c23add6b02a42172c82","subject":"Heat notifier: notify alarm without Octavia tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e2d2ebf44ebb0e5120566c23add6b02a42172c82"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-27 09:16:56.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-27 09:16:56.000000000","tz":480},"subject":"Add the native zuulv3 jobs instead of legacy way","message":"Add the native zuulv3 jobs instead of legacy way\n\nChange-Id: I47d1753f2700c1339171c471a682718a5b72c9d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/402b5c77a9ad2e1221d96eb73ffaec79d2e50c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/402b5c77a9ad2e1221d96eb73ffaec79d2e50c2b"}]},"branch":"refs/heads/master"},"aac850152c566330a63f8870119b62d49eb6d9e4":{"kind":"REWORK","_number":2,"created":"2019-08-27 11:01:45.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/62/678762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/62/678762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/678762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/678762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/678762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/62/678762/2"}}},"commit":{"parents":[{"commit":"e2d2ebf44ebb0e5120566c23add6b02a42172c82","subject":"Heat notifier: notify alarm without Octavia tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e2d2ebf44ebb0e5120566c23add6b02a42172c82"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-27 09:16:56.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-27 11:01:39.000000000","tz":0},"subject":"Add the native zuulv3 jobs instead of legacy way","message":"Add the native zuulv3 jobs instead of legacy way\n\nChange-Id: I47d1753f2700c1339171c471a682718a5b72c9d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/aac850152c566330a63f8870119b62d49eb6d9e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/aac850152c566330a63f8870119b62d49eb6d9e4"}]},"branch":"refs/heads/master"},"bc64b28a04ebdf7070e7297f7bce13e08425af31":{"kind":"REWORK","_number":3,"created":"2019-08-28 00:59:25.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/62/678762/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/62/678762/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/678762/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/678762/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/62/678762/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/62/678762/3"}}},"commit":{"parents":[{"commit":"e2d2ebf44ebb0e5120566c23add6b02a42172c82","subject":"Heat notifier: notify alarm without Octavia tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e2d2ebf44ebb0e5120566c23add6b02a42172c82"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-27 09:16:56.000000000","tz":480},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2019-08-28 00:59:05.000000000","tz":480},"subject":"Add the native zuulv3 jobs instead of legacy way","message":"Add the native zuulv3 jobs instead of legacy way\n\nChange-Id: I47d1753f2700c1339171c471a682718a5b72c9d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bc64b28a04ebdf7070e7297f7bce13e08425af31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bc64b28a04ebdf7070e7297f7bce13e08425af31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
