)]}'
{"id":"openstack%2Faodh~681196","triplet_id":"openstack%2Faodh~master~I8b0b07bdbe7145fb080e77726e76db5108a2c160","project":"openstack/aodh","branch":"master","topic":"eyalb/jsonschema","hashtags":[],"change_id":"I8b0b07bdbe7145fb080e77726e76db5108a2c160","subject":"Update jsonschema according to requirements","status":"MERGED","created":"2019-09-10 09:48:15.000000000","updated":"2019-09-10 22:36:20.000000000","submitted":"2019-09-10 22:35:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681196-1568154901919-91ac400b","meta_rev_id":"c302b57f4c77ad0832f4a3dc705470ad0b0c8d05","_number":681196,"virtual_id_number":681196,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":2,"date":"2019-09-10 22:35:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2019-09-10 21:36:48.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2019-09-10 21:36:48.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-10 21:36:48.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2019-09-10 22:35:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"155386d7455441325b7a1e1d07ab5675db6b216e","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2019-09-10 09:48:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3027bfd89701506b6f32fe9eebe32cdbcc140f59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 10:42:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ffd63ede0f04af3b7fd11457576511c : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9194ee30621f4432ad355c14ef77c5de : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac3162867674e5481ff177f1b8858ec : SUCCESS in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/492005c81d1b44c3a4be1e5e8aebd316 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae8147f558914dab98e839c64dc7fa22 : SUCCESS in 6m 21s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4a35bcac812d47efa7c7b84a95e5d767 : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":1},{"id":"6575a8310f6f193467f21d8304bc3abd00f4b66b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-09-10 21:36:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9590ee7e35ed6034c537c6428c65b00e2393809f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 21:37:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c94afc768c56b19fe47b29c125376d7636507060","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 22:35:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b7226efe1b49dd8b8bb1241eeca2dc : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afd66d80d6d54cf195fcd2cf9e61dfad : SUCCESS in 10m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/346b983e6d834a6ea894567b3b8dfa01 : SUCCESS in 6m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f4936edfb8ab4a2fa111f5b8cf9e86ee : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ecf9ffec5d4cbabf49c75eef629f91 : SUCCESS in 4m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/09704852635f4fa5bdff1eb7047a5a35 : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":1},{"id":"48fb758019545516182cb47ce5fe461bb392c133","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 22:35:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"84801c8b769d9e0b6dd23642f2e3cf1303e6620a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 22:36:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b483b9d32e40179a431ae9f3c5db97 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c79990e8a35137b621112074094d708b75d982d6","revisions":{"c79990e8a35137b621112074094d708b75d982d6":{"kind":"REWORK","_number":1,"created":"2019-09-10 09:48:15.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/96/681196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/96/681196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/681196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/681196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/96/681196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/96/681196/1"}}},"commit":{"parents":[{"commit":"506a99a6f841893c8e785368b11f0e67b11f9836","subject":"Run \u0027telemetry-dsvm-integration-ipv6-only\u0027 job in gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/506a99a6f841893c8e785368b11f0e67b11f9836"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-09-10 09:47:35.000000000","tz":180},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2019-09-10 09:47:35.000000000","tz":180},"subject":"Update jsonschema according to requirements","message":"Update jsonschema according to requirements\n\nChange-Id: I8b0b07bdbe7145fb080e77726e76db5108a2c160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c79990e8a35137b621112074094d708b75d982d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/c79990e8a35137b621112074094d708b75d982d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
