)]}'
{"id":"openstack%2Faodh~687523","triplet_id":"openstack%2Faodh~master~I30cfe9d1cc8c7209e6f865bdd721494dcb4e33ed","project":"openstack/aodh","branch":"master","topic":"fix-get-unhealthy-resource","hashtags":[],"change_id":"I30cfe9d1cc8c7209e6f865bdd721494dcb4e33ed","subject":"heat notifier: Fix getting unhealthy resource names","status":"MERGED","created":"2019-10-09 09:59:00.000000000","updated":"2019-10-10 03:37:31.000000000","submitted":"2019-10-10 03:30:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687523-1570678211168-8e81ebfc","meta_rev_id":"fa1cab261476544d30915c5d572f19396231d944","_number":687523,"virtual_id_number":687523,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 03:30:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-10-10 00:53:56.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"date":"2019-10-09 21:48:54.000000000","_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":1,"date":"2019-10-10 02:35:23.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 21:48:54.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2019-10-10 00:53:56.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-10-10 03:30:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6a0d4944427b2b4749c57048e3d5ae11c4beaf3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-09 09:59:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1eafd27742593c14164ac15f588fbbdc89ac3d1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 11:06:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f01f4d5176c349e58cd47e40ba25083d : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4684c23111e84b2ba9a52b23a336a2cd : SUCCESS in 9m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1924936b4fdf4889a0b79f21d4e7cec9 : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95844dbdab1941268a8ca3c95602362c : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14b93cae0aa2432da10de63ea13b1cbd : SUCCESS in 8m 04s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d69a43db3e5748e4bae2e06feed24b64 : SUCCESS in 52m 23s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/76df7d81279c44d4885e42e610213abf : SUCCESS in 45m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8a9d89b6b2b64fa483e48d2fedeaddf7 : SUCCESS in 43m 56s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2cdb7539a0834396b9cc8e1e1ef20ac5 : SUCCESS in 37m 39s","accounts_in_message":[],"_revision_number":1},{"id":"eba4e5e6a9a51d70a4baccad59e51be8effb4e5e","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-10-10 00:53:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf37bf236748fa865fdf74bbeb54008703bde346","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-10-10 00:55:57.000000000","message":"Patch Set 1:\n\nHi, Lingxian, should we backport this to stable/train?","accounts_in_message":[],"_revision_number":1},{"id":"8e323a99403190eb524012794b4711b91e55a10e","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-10 02:35:23.000000000","message":"Patch Set 1: Workflow+1\n\n\u003e Hi, Lingxian, should we backport this to stable/train?\n\nYeah, thanks for the reminder, I can do that once it\u0027s merged. I will merge it now because I\u0027ll be on holiday from next week.","accounts_in_message":[],"_revision_number":1},{"id":"234ea0adddab1a8319abe50b9e23303e99121bf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 02:35:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e587172cc052813158d32f7d212a671769855755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 03:30:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/881a04de934046139d74765bc4390eff : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0e2dfc7dc27425e9afd6b1943d94f21 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e0de833fe1a46948f0e1cf00e98f60f : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a9e9cc2ed2b4e9ba3a599029dab2e3d : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48069d28fd44424b813ed76ded4fe1f9 : SUCCESS in 5m 35s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/cbfee7f685324620a0b4fc469776fd9d : SUCCESS in 49m 12s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/450746d512694c3d9e996747e993d503 : SUCCESS in 46m 55s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6377c94389f745e795367bb5d559a0a2 : SUCCESS in 48m 10s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/12fa84e450834e0d9e051c394d64bd4c : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c0f072e2c3157175ea95dc4668053bcfe5dc0334","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 03:30:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8fc8f48fac08da21a5b7a079ec277b79c7433a1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 03:31:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ad01551c5484b34ae56459ab21b7723 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0b207e283ebdc0ae4a7b3d5418382169d93e535d","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-10-10 03:37:31.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 9e17daad69c65792c8e03d6e39a9601f6972a8c6","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b47550ae447c81135a1145c7ea8b90004798b4b5","revisions":{"b47550ae447c81135a1145c7ea8b90004798b4b5":{"kind":"REWORK","_number":1,"created":"2019-10-09 09:59:00.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/23/687523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/23/687523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/23/687523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/23/687523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/23/687523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/23/687523/1"}}},"commit":{"parents":[{"commit":"68f1effa10176a3a08f5ad9fe357e73bb4f33e4f","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/68f1effa10176a3a08f5ad9fe357e73bb4f33e4f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-09 09:55:44.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-10-09 09:58:08.000000000","tz":780},"subject":"heat notifier: Fix getting unhealthy resource names","message":"heat notifier: Fix getting unhealthy resource names\n\nIn the autoscaling scenario, the resource which should be marked as\nunhealthy is always the parent resource of the failed load balancer pool\nmember.\n\nChange-Id: I30cfe9d1cc8c7209e6f865bdd721494dcb4e33ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b47550ae447c81135a1145c7ea8b90004798b4b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b47550ae447c81135a1145c7ea8b90004798b4b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
