)]}'
{"id":"openstack%2Faodh~700089","triplet_id":"openstack%2Faodh~master~I024ff3ec0e5c21f2f6cef9fa453cfb1497921947","project":"openstack/aodh","branch":"master","topic":"fix-get-alarms","hashtags":[],"change_id":"I024ff3ec0e5c21f2f6cef9fa453cfb1497921947","subject":"Fix getting alarms","status":"MERGED","created":"2019-12-20 00:46:30.000000000","updated":"2019-12-20 10:29:55.000000000","submitted":"2019-12-20 10:28:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700089-1576837704899-fead5b33","meta_rev_id":"2eb2b0ee7e4f5e5dc62cc697b1f9386769e16b67","_number":700089,"virtual_id_number":700089,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 10:28:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2019-12-20 09:24:26.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2019-12-20 09:24:26.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 09:24:26.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-12-20 10:28:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"75bf066c544e34f63298fd11b8109978227cf5b0","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-20 00:46:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b7ee793d81f624c1ffe94e251833436de51a81d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 01:43:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9c465625278497098e69504a6473311 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3e8580a40534c879fc59fb656282d2d : FAILURE in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87ef2c4e67a74658ab85640f1e02cf81 : FAILURE in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ca21a848c24b978d58591c213e1f6b : SUCCESS in 6m 00s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/48bf9a80444747b8be722a34ebdbb758 : SUCCESS in 46m 34s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/72f5a09d0e5d41998241816f5cb014dc : SUCCESS in 41m 34s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2cfec267d8a944f4ae4684311168124b : SUCCESS in 50m 44s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/529938d1bed44ad8b0692ed63f2c3a96 : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":1},{"id":"853c02c958f581296f1d78639ea2551add99b68f","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-20 03:06:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc46a24cacd61d43edc7c0e81d5110d80e676d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 04:07:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a32a19e24243bca6fba2d465d71db9 : SUCCESS in 11m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8f51e1e996342838bcde94c6a384e66 : SUCCESS in 13m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42828474e9e14d69b4c43ff7a11bc87c : SUCCESS in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34eb0e3ced3542abaa667ff402049442 : SUCCESS in 14m 02s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6b7f4a7d7fcf40939e8ee1f35212da40 : SUCCESS in 52m 23s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/95c39066806849bf8d5e8043d37de4f5 : SUCCESS in 42m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/30db2d216d494858b233954a0b85631c : SUCCESS in 51m 50s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/af8fb4f365e940b28dd723258443fd44 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":2},{"id":"2a4be05257bd899bcf53e0c321e4e192d87b9010","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2019-12-20 09:24:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0dccfde46569508db2606578a3211ac88089fccc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 09:24:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0e7170bb8ed3e2fb6bc374fbe472033cdba60ac5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:28:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e9e85039d9245df971daa586cf9ea6e : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b1242b8c1e64aee839f82c1adeda0ba : SUCCESS in 7m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4be0dac32b44a9a9c284c237d57e231 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26583547e424ffa8462e271785445a2 : SUCCESS in 6m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/49051f9dbf2b472c9a3e0485f3292fbe : SUCCESS in 47m 38s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5a6c3cddf62b47a485affded53047bbd : SUCCESS in 55m 24s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/07a7c467cb304edfbe24deaad09b8d95 : SUCCESS in 1h 02m 06s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9a2fc9a7288d489caf4b20a28f38c452 : SUCCESS in 46m 18s","accounts_in_message":[],"_revision_number":2},{"id":"66e7179e5badc99e6f74aa848af97468ade9c26b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:28:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"794b52b208285cc6eb535c6daef672d7ab465ccf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:29:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91da85df76b54d9ba26d7d35b2eb7ab4 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b76f377781051fa75117ac7101b1a68be3575390","revisions":{"f0956157e21b57fc14f25ae709eca6d14ad092f2":{"kind":"REWORK","_number":1,"created":"2019-12-20 00:46:30.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/89/700089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/89/700089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/700089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/700089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/700089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/89/700089/1"}}},"commit":{"parents":[{"commit":"b18d0bae7c48377db91961ead05dbae92dfb8c37","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b18d0bae7c48377db91961ead05dbae92dfb8c37"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-20 00:45:16.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-20 00:45:18.000000000","tz":780},"subject":"Fix getting alarms","message":"Fix getting alarms\n\nThe normal users can get all other projects alarms because no project_id\nfilter when querying the db.\n\nChange-Id: I024ff3ec0e5c21f2f6cef9fa453cfb1497921947\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f0956157e21b57fc14f25ae709eca6d14ad092f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f0956157e21b57fc14f25ae709eca6d14ad092f2"}]},"branch":"refs/heads/master"},"b76f377781051fa75117ac7101b1a68be3575390":{"kind":"REWORK","_number":2,"created":"2019-12-20 03:06:30.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/89/700089/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/89/700089/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/700089/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/700089/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/89/700089/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/89/700089/2"}}},"commit":{"parents":[{"commit":"b18d0bae7c48377db91961ead05dbae92dfb8c37","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b18d0bae7c48377db91961ead05dbae92dfb8c37"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-20 00:45:16.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-20 03:04:57.000000000","tz":780},"subject":"Fix getting alarms","message":"Fix getting alarms\n\nThe normal users can get all other projects alarms because no project_id\nfilter when querying the db.\n\nChange-Id: I024ff3ec0e5c21f2f6cef9fa453cfb1497921947\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b76f377781051fa75117ac7101b1a68be3575390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b76f377781051fa75117ac7101b1a68be3575390"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
