)]}'
{"id":"openstack%2Faodh~704751","triplet_id":"openstack%2Faodh~master~I9258ae4801edbc5289d890fe2e060964a73b216c","project":"openstack/aodh","branch":"master","topic":"quota-api","hashtags":[],"change_id":"I9258ae4801edbc5289d890fe2e060964a73b216c","subject":"Support quota API","status":"MERGED","created":"2020-01-29 09:09:49.000000000","updated":"2020-01-30 05:28:54.000000000","submitted":"2020-01-30 05:27:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704751-1580362060085-641ca554","meta_rev_id":"f2c276c3d44d1e917ebaa35626379c57e37df984","_number":704751,"virtual_id_number":704751,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 05:27:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":2,"date":"2020-01-30 04:34:02.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"date":"2020-01-29 10:30:00.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":1,"date":"2020-01-30 04:34:02.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 10:30:00.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-01-30 05:27:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0b722c8b9ddd0b6b446060d9e713f5d311b36bb","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-29 09:09:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69b9e961528d1a106af8fd2fa338b7fd78ea1003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 10:03:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f17a4440b4493db6ce4f044c6c7de6 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca285a2e31f44e5ab38b0920037076ed : SUCCESS in 10m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d331140bc5404e838aee0ae2dc0e5564 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb6d1008cda478f873caa6e7dd44e14 : FAILURE in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/882e88161f2f4704a2ef586b0bcef757 : SUCCESS in 7m 25s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/7d1560f40b784a4f9aef3372ea009a69 : SUCCESS in 38m 29s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/929afb069e794facbfe0485896da0405 : SUCCESS in 46m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/06e9a8ed46fe4920a4126842ac1e6b56 : SUCCESS in 44m 18s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d444b6aee5204fe1a59d408c9e499a1e : SUCCESS in 47m 31s","accounts_in_message":[],"_revision_number":1},{"id":"72f93314c2403cd62286e67944b57cfac94d5b7c","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-30 04:34:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d18d2c4a833e0b8b5ba813666bf7400e4310901","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 04:34:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e0eaf712d3b3edd059b504ef60f6f2ee56f6c5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 05:27:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edb3c98ec29143ef9f1bdee5a7fef50c : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87d27490d19f411b8845dd0770d24861 : SUCCESS in 8m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fc1be43677b47e0a4e80f979e333d90 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b89c72391646cfbd6cbbaab94c114c : SUCCESS in 8m 21s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e7e6f86cf39445f687211b1b4bc7d7f5 : SUCCESS in 41m 22s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3b6a2441cfab464fa663ce24208567b9 : SUCCESS in 41m 20s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2b05b04c37ca4a98890a0b375ff35612 : SUCCESS in 46m 17s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/edbc93cfe5eb41e5a3e31c5b4c8b29e1 : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c8d0ee645a2c7d64f7e9119418ee78c79dda468c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 05:27:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f52dd8fa80314b0f07303e4977d22292af0da90c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 05:28:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a3aa1b8f3e42b1a761c72b0d25e993 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"223178716e784aa1e65c656dc639a18ddf4fba9f","revisions":{"223178716e784aa1e65c656dc639a18ddf4fba9f":{"kind":"REWORK","_number":1,"created":"2020-01-29 09:09:49.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/51/704751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/51/704751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/51/704751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/51/704751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/51/704751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/51/704751/1"}}},"commit":{"parents":[{"commit":"58dd647097d47064dbe4210b3e9f44584d1b085b","subject":"Use config options when creating keystone session","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/58dd647097d47064dbe4210b3e9f44584d1b085b"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-28 05:28:35.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-29 09:09:39.000000000","tz":780},"subject":"Support quota API","message":"Support quota API\n\nCurrently Aodh only supports global quota for projects and users, we\nneed quota API to manage quotas per project.\n\nThe new quota API only supports project quotas as the same with other\nopenstack services.\n\nBy default, only admin user can update the quota.\n\nTODO(in subsequent patches):\n- Check quota from DB when creating resources\n- Doc\n- Release note\n- openstack CLI support\n\nChange-Id: I9258ae4801edbc5289d890fe2e060964a73b216c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/223178716e784aa1e65c656dc639a18ddf4fba9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/223178716e784aa1e65c656dc639a18ddf4fba9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
