)]}'
{"id":"openstack%2Faodh~705130","triplet_id":"openstack%2Faodh~master~I1af381f357b0d80f68e3ead16e158428b7a14555","project":"openstack/aodh","branch":"master","topic":"check-quota","hashtags":[],"change_id":"I1af381f357b0d80f68e3ead16e158428b7a14555","subject":"Improve the quota check","status":"MERGED","created":"2020-01-30 22:57:11.000000000","updated":"2020-01-31 02:05:08.000000000","submitted":"2020-01-31 02:03:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705130-1580436214754-a46f390a","meta_rev_id":"06e01f2af42c14f96c9230a3d2de89f4b3fc1782","_number":705130,"virtual_id_number":705130,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-31 02:03:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-31 01:17:36.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-31 01:17:36.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 02:03:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a20fd310bff5b1f593562e5271f5e25af8c47f3a","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-30 22:57:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e232933e39b17d90822f8f276cd370c45d96c569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 23:54:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee9038c9d7ab41b09865aec7aa483ceb : FAILURE in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a24d9a691244076bdde5d89cef0e343 : SUCCESS in 11m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b577cdaebea64495a1762a7804172d7a : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12f0f4655aac4e03ad90f9c76a753278 : FAILURE in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c78f8e3b029f42ebb18003fb77b0cfc2 : SUCCESS in 11m 03s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/8c192adfbaab4c8181fcdd6825f58237 : SUCCESS in 50m 52s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/fb0d0541cd504c3fb284c4de055dbcf2 : SUCCESS in 47m 47s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cb9c001fc63f408c8a22f2ce2771222f : SUCCESS in 54m 00s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/a4f68e24d7f640889899a71ef4fa7218 : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":1},{"id":"716a6dd26e048d90c4d8ca198b5b162d0eed4667","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-31 00:19:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"afce88f6f112bdcdde095a898e6dd87fd8005576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 01:15:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06a75a32da9461db4807fb6c6667f5e : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68d875f1ccea4e5e9612ff1bb1abc1b1 : SUCCESS in 15m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7bc664290f64e31864974e4693ddfd0 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9c549c85e964fb6a69cc0ea815619de : FAILURE in 13m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a2cf22675e4f2d9a33e3f560d63334 : SUCCESS in 5m 53s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/5a7c8d9fc36d47338cd0867d90a6b620 : SUCCESS in 47m 44s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/9e8ef17cd6c64720b7089d68c989bf77 : SUCCESS in 42m 36s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d4568907a0034013a12e7d8dc8169667 : SUCCESS in 48m 30s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d67e089c7fed4d198c91eb83ab891cf7 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":2},{"id":"cd56f3cc8066d63b1b4e59db41e1a8385898a037","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-31 01:17:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5f40931da6a1abe56b0592415791a4db9c3bd075","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 01:17:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b199c7eed09b92d7a5a9737b731c8495c8b9a4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 02:03:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66e4e41841144564b11725fd564fb446 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb7559b3d55340febe774861915910a6 : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adcc2d99d13045919a65ca8426082cd4 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d134a87abe846218626e73b6be86eb0 : SUCCESS in 8m 17s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/5bc1a19c01a84a8fb11ff013498fa9cc : SUCCESS in 39m 32s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8bb836925e7d4223bc6485fd9b819f85 : SUCCESS in 39m 13s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bfdb3553ef6b4867a13f0a746dee998a : SUCCESS in 45m 07s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2a724714a8df4706977aa56747a056ac : SUCCESS in 43m 54s","accounts_in_message":[],"_revision_number":2},{"id":"b26dab3c94f33cce5ef1591bf4f3276b1cf8cc85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 02:03:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f6ae4ea91f65f17bc6130460b6064ca6c33c6d5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 02:05:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b6932e2ab34020aada363cfa3cfaff : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"90a7c6eab2c897bb613bb3aa489f4c075370138a","revisions":{"99b0b9b1b83589421e34b599162c883b8c2b6c53":{"kind":"REWORK","_number":1,"created":"2020-01-30 22:57:11.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/30/705130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/30/705130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/30/705130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/30/705130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/30/705130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/30/705130/1"}}},"commit":{"parents":[{"commit":"223178716e784aa1e65c656dc639a18ddf4fba9f","subject":"Support quota API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/223178716e784aa1e65c656dc639a18ddf4fba9f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-30 21:02:51.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-30 22:56:08.000000000","tz":780},"subject":"Improve the quota check","message":"Improve the quota check\n\n- Check alarm quota by considering the quota setting in DB.\n- Add more unit tests for quota API.\n\nChange-Id: I1af381f357b0d80f68e3ead16e158428b7a14555\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/99b0b9b1b83589421e34b599162c883b8c2b6c53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/99b0b9b1b83589421e34b599162c883b8c2b6c53"}]},"branch":"refs/heads/master"},"90a7c6eab2c897bb613bb3aa489f4c075370138a":{"kind":"REWORK","_number":2,"created":"2020-01-31 00:19:43.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/30/705130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/30/705130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/30/705130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/30/705130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/30/705130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/30/705130/2"}}},"commit":{"parents":[{"commit":"223178716e784aa1e65c656dc639a18ddf4fba9f","subject":"Support quota API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/223178716e784aa1e65c656dc639a18ddf4fba9f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-30 21:02:51.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2020-01-31 00:19:33.000000000","tz":780},"subject":"Improve the quota check","message":"Improve the quota check\n\n- Check alarm quota by considering the quota setting in DB.\n- Add more unit tests for quota API.\n\nChange-Id: I1af381f357b0d80f68e3ead16e158428b7a14555\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/90a7c6eab2c897bb613bb3aa489f4c075370138a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/90a7c6eab2c897bb613bb3aa489f4c075370138a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
