)]}'
{"id":"openstack%2Faodh~719154","triplet_id":"openstack%2Faodh~master~I69f42c50f94d9511179185b6f0de37a8b8dd0469","project":"openstack/aodh","branch":"master","hashtags":[],"change_id":"I69f42c50f94d9511179185b6f0de37a8b8dd0469","subject":"Cleanup py27 support","status":"ABANDONED","created":"2020-04-11 08:03:30.000000000","updated":"2020-04-13 07:51:35.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eeea8b3f2b01fe3b0571c4c1b554a0a2f27c8604","_number":719154,"virtual_id_number":719154,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-13 06:41:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 06:41:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfa484c2d053f046541e0aae26564e894064b781","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-11 08:03:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaff15c2fceb177cc28c138c4de6197f857bbf07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 08:48:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1192919e8f094ed3ae1a655d6710bc9a : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f52bb4995e64e57a4275f118b50925a : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d3c1ba2c1ef4d5ca0b3c8a6dc3fc90c : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d207ace058c245d4a47795bd77c06bba : SUCCESS in 21m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1047023aaaf84fcead02164333b469a4 : FAILURE in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e516642efd47cab59b233e9ec96794 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4546bb2a99b6486794536b89ca84509c : SUCCESS in 9m 33s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4c85b2fc0b1a442bb7c979d021b5cef9 : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"cbe0b05dda8d44ec048adfd2c687a16b2274f90e","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 00:54:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4fda871d966f6d828bd925e27b49e207bcc7d4a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 00:54:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9e514506f9c923ba9946afdfb034fc84a595072d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 01:43:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f6f7a9cbd38e4c2084abb711a260c8d2 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9595e80a9c4436e9f98e43653eb1d9a : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac48cfd86c7a4421b3568974ea245fea : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5775fe6e02c54a19a07e0a82c269ec5f : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a89b6f6f908c4781934ac3f13019f509 : FAILURE in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dda1568bfe9477a9749feed2a64a3e7 : SUCCESS in 8m 25s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c97717a3cb47410399dcf22df69db90d : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":1},{"id":"141a04c9d055912834856f835b2f043615c439a4","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 02:35:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4934a825120d3e7ea5c013b255b00c45d81ae945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 03:20:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/27cd688f83b444cab67c2647167d7405 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e30c7a7f15642778da35d02251fcba5 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6b1575500054c29a703a587cc7e449b : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e860b3478b054bcb9e35bb07d5d3fdb7 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1ab9b66cbe04587a12042cd73112423 : FAILURE in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9b30eff2444936a7ef272741212c44 : FAILURE in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7d376220c274522807c871af5a9baa3 : SUCCESS in 8m 53s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/35819888c1f64e9d86ec1b64dce3b897 : SUCCESS in 43m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f55efbe11036e455c09f263902717a594cea24aa","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 03:21:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"35e548d2fd963b59b23f7924d36476de833fc2c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 04:11:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ba812688f501493c9e62a45b03c95fd6 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a3d44ba3a0f42da91983b3a512e15f5 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5733d50f245a4137ae3ba00b198f135c : SUCCESS in 11m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/76478325b38a49bb817ac8eea124e70d : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13bfeb6401834ca09a6afd735a595fb7 : FAILURE in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133660a336054f6c96838989e55bc13d : FAILURE in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fb6183a97fd463a96c748abb5785f88 : SUCCESS in 9m 03s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/8a7341c5269d44ed805d641fc51761c5 : SUCCESS in 44m 14s","accounts_in_message":[],"_revision_number":3},{"id":"4b8a0fb5c8a87fe33f42e2c21c6fcb39b0e6568e","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 05:50:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec3cd29df6c21d355d24a04bb87a314f7ed1e846","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 06:41:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f7b4b104223d40e5a5c7a7c2d06e1f87 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f20d9c5eb0a494faaab7cae82349b97 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6def178262a4819bc2adb578498e973 : SUCCESS in 9m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0dd088c99a5453faa0f78a5e6d411f3 : SUCCESS in 9m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/196c12b0193247e5af37414a32de369e : FAILURE in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d939b680f7c64cb2b6b8a68294067770 : FAILURE in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ecd62cca6484b0aa715ae2114f02836 : SUCCESS in 8m 56s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0320838bba254c72ba285f4b125dbef9 : SUCCESS in 46m 44s\n\nWarning:\n  Comments left for invalid file doc/source/contributor/webapi/v2.rst","accounts_in_message":[],"_revision_number":4},{"id":"eeea8b3f2b01fe3b0571c4c1b554a0a2f27c8604","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 07:51:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a1d3d4c4a4101cfd52065837659fc67137dfa020","revisions":{"08d07b03166f205dd0e6d32510597248ee5a85f6":{"kind":"REWORK","_number":1,"created":"2020-04-11 08:03:30.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/54/719154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/54/719154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/54/719154/1"}}},"commit":{"parents":[{"commit":"6ae7f90ad272e0beb1da028156ed6b7a19c8c880","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6ae7f90ad272e0beb1da028156ed6b7a19c8c880"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-11 08:02:47.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-11 08:02:47.000000000","tz":-420},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg:\n  * Wheel is not needed for python 3 only repo\n  * Some other sections are obsolete\n- Update classifiers\n- Update requirements, no need for python_version anymore\n\nChange-Id: I69f42c50f94d9511179185b6f0de37a8b8dd0469\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/08d07b03166f205dd0e6d32510597248ee5a85f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/08d07b03166f205dd0e6d32510597248ee5a85f6"}]},"branch":"refs/heads/master"},"61bb40cc2a5d82df3903a4bcc7720a17a4be258a":{"kind":"REWORK","_number":2,"created":"2020-04-13 02:35:55.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/54/719154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/54/719154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/54/719154/2"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-11 08:02:47.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 02:35:29.000000000","tz":-420},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg:\n  * Wheel is not needed for python 3 only repo\n  * Some other sections are obsolete\n- Update classifiers\n- Update requirements, no need for python_version anymore\n\nChange-Id: I69f42c50f94d9511179185b6f0de37a8b8dd0469\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/61bb40cc2a5d82df3903a4bcc7720a17a4be258a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/61bb40cc2a5d82df3903a4bcc7720a17a4be258a"}]},"branch":"refs/heads/master"},"174b3835d7b1b4932d11a3b2040b608790ed372c":{"kind":"REWORK","_number":3,"created":"2020-04-13 03:21:53.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/54/719154/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/54/719154/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/54/719154/3"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-11 08:02:47.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 03:21:33.000000000","tz":-420},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg:\n  * Wheel is not needed for python 3 only repo\n  * Some other sections are obsolete\n- Update classifiers\n- Update requirements, no need for python_version anymore\n\nChange-Id: I69f42c50f94d9511179185b6f0de37a8b8dd0469\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/174b3835d7b1b4932d11a3b2040b608790ed372c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/174b3835d7b1b4932d11a3b2040b608790ed372c"}]},"branch":"refs/heads/master"},"a1d3d4c4a4101cfd52065837659fc67137dfa020":{"kind":"REWORK","_number":4,"created":"2020-04-13 05:50:31.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/54/719154/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/54/719154/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/54/719154/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/54/719154/4"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-11 08:02:47.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 05:50:11.000000000","tz":-420},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nMake a few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg:\n  * Wheel is not needed for python 3 only repo\n  * Some other sections are obsolete\n- Update classifiers\n- Update requirements, no need for python_version anymore\n\nChange-Id: I69f42c50f94d9511179185b6f0de37a8b8dd0469\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a1d3d4c4a4101cfd52065837659fc67137dfa020"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a1d3d4c4a4101cfd52065837659fc67137dfa020"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
