)]}'
{"id":"openstack%2Faodh~719395","triplet_id":"openstack%2Faodh~master~Ibb42f8199eb4ca93e438e5e7ef5d4c4e0b35a738","project":"openstack/aodh","branch":"master","hashtags":[],"change_id":"Ibb42f8199eb4ca93e438e5e7ef5d4c4e0b35a738","subject":"Cleanup py27 support","status":"MERGED","created":"2020-04-13 06:06:27.000000000","updated":"2020-04-14 03:42:16.000000000","submitted":"2020-04-14 03:40:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719395-1586835653099-d1a6a7d1","meta_rev_id":"eb0d6e54e426db8e0338b0edc723294dd368e4f0","_number":719395,"virtual_id_number":719395,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 03:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-04-14 00:40:33.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-04-14 00:40:33.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 00:40:33.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-04-14 03:40:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bd0e3aee253c61e3dd2ec73da4d4bcbc16605fce","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 06:06:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64cec6656cac0cea35c015710cde1285b602e0bc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 06:07:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"716c444b860692ba9c24bbba4ece088721fd7c48","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 06:12:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d3dee94dbe37ff80ce4bd904280e6c5c29dae542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 07:11:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/967f2e3945ca44cab3bb37ac69f482ac : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5675748b88d94696b37af9a2114a9ff5 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bada8d2bd6c64034a592bce31a55b444 : SUCCESS in 9m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6790df5a29bd4e33a186ebce910b1fdf : SUCCESS in 11m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ec961f10c88431ab82e95dc97a665f2 : FAILURE in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5c02bf1fbf490fb8dec20136462259 : FAILURE in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bf3bd711ff64cab8fa13561173eac07 : SUCCESS in 11m 47s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2b45344230c04a00b6443de26e0d161d : SUCCESS in 44m 38s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/047d940f2b174f0b92b73809e083fcd6 : SUCCESS in 45m 20s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/363d9dd49b2b42fd8bd98cd67ed641bc : SUCCESS in 48m 46s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0768ac5dfaaf4e70b58ac2d06ffad82a : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":3},{"id":"aad5d537f27030e0bb11560e06c6dc422c6b66ff","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 08:59:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5a1f5b3c1ed9cce4ed41c672e93fa2339c989e80","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 09:12:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3459a15259a0638f1ed3ed17fc515007ab1ff75c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:52:36.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c6b2d8f8cde74fa785430329ec4bf220 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d2f97f76bf94f1db91e3f379ccae956 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2714de081b244c37adb69d8ad1e7e4dc : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd56595840f6451496a7a3e5e8794a51 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab892f2634d44f45b5e8ec3cedcf74e0 : FAILURE in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f0836da5e1414b990b084b4b3c6d10 : FAILURE in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7480876ac9784c04bed769348075fbf5 : SUCCESS in 9m 21s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/fd9208a9d9f84789b095d4713aeced38 : SUCCESS in 39m 07s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/603b4461603144f7a80439b19166e36c : SUCCESS in 40m 18s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a19a5df4dc37493e92873bb701df2c0a : SUCCESS in 45m 10s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/eacd5bbc305e4b6faef05d2ac6e9c795 : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":4},{"id":"81096c4b332f7ca02a8f476bebfa991d512c7352","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 09:54:04.000000000","message":"Patch Set 4: Workflow-1\n\nLet\u0027s wait for wsme fix.","accounts_in_message":[],"_revision_number":4},{"id":"3936d9937d2d4f96bc3d749f5c2ee357bb7e205a","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 10:38:22.000000000","message":"Patch Set 4:\n\n\u003e Let\u0027s wait for wsme fix.\n\nHi AJ, what\u0027s your opinon about my workaround fix.\nhttps://review.opendev.org/#/c/719405/","accounts_in_message":[],"_revision_number":4},{"id":"592555cd131e8a8dac7c971637d87945f27b8d9a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 15:01:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4e08657a85dc3c5711569be9ae9ed51f7095c1b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 15:57:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/43c49534a9d64f3099cb6bf9b3c9e77f : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe26ac50e0524345b94dc11adf1fb3a0 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac5a1ec19a4542b4b78d9c483844b740 : SUCCESS in 7m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7a8cd47657254f6fa2cf02329f5a61e0 : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa54d08971dd4d6baedc14a42494e92a : FAILURE in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74250dc50aef4f0b9fa2d2d5a559e698 : FAILURE in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bdd7182e5524037ae601b6d6434b205 : SUCCESS in 12m 29s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/770e2a4469304af782d672f399069ead : SUCCESS in 41m 29s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/8923ed274e3f42f3915878fde0fcb4ad : SUCCESS in 42m 52s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/87bdc76655f74779a915974f8f5514d1 : SUCCESS in 48m 21s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e588e951281b43348b5f355025130f00 : SUCCESS in 46m 26s\n\nWarning:\n  Comments left for invalid file doc/source/contributor/architecture.rst","accounts_in_message":[],"_revision_number":5},{"id":"3e742dae2ecc767757c9c3985581bf516f78034b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 16:04:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1ab6aa12f51acfcae3208e366ec247ebad51db5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 17:00:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5142efc3036242a7815a57c36f6921f0 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35628dbea854438c9583f146eae21035 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1524897fff14bac8f916d38ab431d1c : SUCCESS in 8m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b273e2fedb3a400a92c13e6109fe1c69 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0761b10c05fd48d8876b58213d47c19a : FAILURE in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6db5bb229f69473f833096b835914dc9 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b5e0433f59048aa8f5f986096d3bfa8 : SUCCESS in 9m 39s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4e927cea72454a41aede180e1eff8523 : SUCCESS in 45m 28s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e7b4af46073a4e2284fafa57ef4bebfd : SUCCESS in 42m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/4794589ca1494147b9cb961a3f6a214b : SUCCESS in 50m 05s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/67de5e897b214b8188edd657e09bd9eb : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":6},{"id":"e559ac5abc7784250eb6dbaf6df28369a3f7ae36","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-14 00:40:33.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"094b8289ab4431613ce7540bb00a030aef10ca1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 00:40:45.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b65a9cecd5ecab58758c1ff5c2127d36528784c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 01:48:43.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/88a88d6d28c643f7b9ac51be21946e2b : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a851168e836c474891099ac819516515 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/414281314de54f0b931d2f8b2e1cf52f : SUCCESS in 15m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07668d83154744e0a92d1e6341d4959c : SUCCESS in 15m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5206a9099ce04d099307acce4e92780f : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d3fa93f45d8483e90bb7a8df0362b1d : SUCCESS in 10m 07s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2e6b0d7fde5242c3bc20261b69db8279 : SUCCESS in 48m 24s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b080546bdb4341bea13549ee9073ae94 : FAILURE in 36m 52s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f52a77fc5ffd4628a0c0fc04a0f96949 : SUCCESS in 44m 41s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/36d64eea3e704cb68a6f3191d8955181 : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":6},{"id":"f72f8cb9c711b9b4c6abb5d98f0b811c94a39327","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-14 02:05:49.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"e8b8c54dbd94d4e99d5a82a55671baf9b385de49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 02:50:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7b8377ee678d40d09eab0dd1deccc8a1 : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afe8107a8b8f49dc9fe594438e58db23 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eecd89b9ed1647ab94b9c6130a0068b5 : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c1388fcfd9a483f9190cd6270d86250 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7402fd263b8c483ba49705f05ad036bb : FAILURE in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c1c18d92f04cdfa62dceccf3747c15 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be365be9aca41b2a82d8c72a6be534f : SUCCESS in 9m 00s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c1eee9de5b974d379be3abebe90420fb : SUCCESS in 40m 18s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/05e6ab4a249645a8a01990f3d55146f7 : SUCCESS in 36m 57s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0fd69b73636244c8b0cc133e9be571aa : SUCCESS in 43m 14s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/cd4af6cade394863876436e3fed7b981 : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":6},{"id":"5069799638fbfee665554526b6c8d49f5c7392d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 02:50:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e0bb05ef88fd8953aed4a61e78300c43749af2c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 03:40:52.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1cbcd655a65649aeb33a7ad25d125b9f : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b45482d9dd44072ad9e9093537d1d3c : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/724ffca2499e4e16b7841cf93e6d53c9 : SUCCESS in 15m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc860f28360a40adaa4812031b34e9d9 : SUCCESS in 16m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be2bbf01f9ff479ab719151ac4583689 : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acf9800fea904ff8b4b4c530b0911339 : SUCCESS in 8m 29s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a3c7c48bfe0f4a6c849de7f9e491288f : SUCCESS in 43m 37s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/87f19f9a1f1d4da0b628cfec84830155 : SUCCESS in 32m 16s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/53f0b29e0ee84a07884bf07df5099561 : SUCCESS in 48m 47s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2fad05cb6b4e4887860e166b0647a0e1 : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":6},{"id":"0a2473457a5b4ca8489c01a757e8f99e29533bca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 03:40:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"eb0d6e54e426db8e0338b0edc723294dd368e4f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 03:42:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45d8f8cf8b9449985d76eee5d18d7f9 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3256e0ba1214b6baedaa65de0533c4f : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"54f2eb0904943d59e54c1ad1d49d96e8c5ac7272","revisions":{"6025c94b21c875f9e45534efd68d9174605db8eb":{"kind":"REWORK","_number":1,"created":"2020-04-13 06:06:27.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/95/719395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/95/719395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/95/719395/1"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 06:01:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 06:05:40.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n- Update requirements, no need for python_version anymore\n- Use newer openstackdocstheme and reno version\n- Update hacking version to 3.0\n\nChange-Id: Ibb42f8199eb4ca93e438e5e7ef5d4c4e0b35a738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6025c94b21c875f9e45534efd68d9174605db8eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6025c94b21c875f9e45534efd68d9174605db8eb"}]},"branch":"refs/heads/master"},"487c4ba18f89a0b89a10f44431f7a67bd8648c8a":{"kind":"REWORK","_number":2,"created":"2020-04-13 06:07:53.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/95/719395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/95/719395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/95/719395/2"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 06:01:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 06:07:47.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n- Update requirements, no need for python_version anymore\n- Use newer openstackdocstheme and reno version\n- Update hacking version to 3.0\n- move python3 setting into toxenv\n\nChange-Id: Ibb42f8199eb4ca93e438e5e7ef5d4c4e0b35a738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/487c4ba18f89a0b89a10f44431f7a67bd8648c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/487c4ba18f89a0b89a10f44431f7a67bd8648c8a"}]},"branch":"refs/heads/master"},"57dcdfaf9d72421481a9166d295672513433860f":{"kind":"REWORK","_number":3,"created":"2020-04-13 06:12:43.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/95/719395/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/95/719395/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/95/719395/3"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 06:01:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 06:12:11.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n- Update requirements, no need for python_version anymore\n- Use newer openstackdocstheme and reno version\n- Update hacking version to 3.0\n- move python3 setting into toxenv, remove install_command and\n  move constraints into deps\n\nChange-Id: Ibb42f8199eb4ca93e438e5e7ef5d4c4e0b35a738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/57dcdfaf9d72421481a9166d295672513433860f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/57dcdfaf9d72421481a9166d295672513433860f"}]},"branch":"refs/heads/master"},"a051ae63bdd2a492d33951edfd690efdff78bbcb":{"kind":"REWORK","_number":4,"created":"2020-04-13 08:59:37.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/95/719395/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/95/719395/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/95/719395/4"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 06:01:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 08:59:30.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n- Update requirements, no need for python_version anymore\n- Use newer openstackdocstheme and reno version\n- Update hacking version to 3.0\n- move python3 setting into toxenv, remove install_command and\n  move constraints into deps\n\nChange-Id: Ibb42f8199eb4ca93e438e5e7ef5d4c4e0b35a738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a051ae63bdd2a492d33951edfd690efdff78bbcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a051ae63bdd2a492d33951edfd690efdff78bbcb"}]},"branch":"refs/heads/master"},"ee7637a4d54861b1b06bc5d175caa168e37e2b5d":{"kind":"REWORK","_number":5,"created":"2020-04-13 15:01:24.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/95/719395/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/95/719395/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/95/719395/5"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 06:01:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 15:01:19.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n- Update requirements, no need for python_version anymore\n- Use newer openstackdocstheme and reno version\n- Update hacking version to 3.0\n- move python3 setting into toxenv, remove install_command and\n  move constraints into deps\n\nChange-Id: Ibb42f8199eb4ca93e438e5e7ef5d4c4e0b35a738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ee7637a4d54861b1b06bc5d175caa168e37e2b5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/ee7637a4d54861b1b06bc5d175caa168e37e2b5d"}]},"branch":"refs/heads/master"},"54f2eb0904943d59e54c1ad1d49d96e8c5ac7272":{"kind":"REWORK","_number":6,"created":"2020-04-13 16:04:46.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/95/719395/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/95/719395/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/719395/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/95/719395/6"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 06:01:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-04-13 16:04:34.000000000","tz":120},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nThis repo is now testing only with Python 3, so let\u0027s make\na few cleanups:\n- Remove python 2.7 stanza from setup.py\n- Add requires on python \u003e\u003d 3.6 to setup.cfg so that pypi and pip\n  know about the requirement\n- Remove obsolete sections from setup.cfg\n- Update classifiers\n- Update requirements, no need for python_version anymore\n- Use newer openstackdocstheme and reno version\n- Update hacking version to 3.0\n- move python3 setting into toxenv, remove install_command and\n  move constraints into deps\n\nChange-Id: Ibb42f8199eb4ca93e438e5e7ef5d4c4e0b35a738\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/54f2eb0904943d59e54c1ad1d49d96e8c5ac7272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/54f2eb0904943d59e54c1ad1d49d96e8c5ac7272"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
