)]}'
{"id":"openstack%2Faodh~719405","triplet_id":"openstack%2Faodh~master~Ibc396a191cb43111b823968c07dbecfab7ea3664","project":"openstack/aodh","branch":"master","hashtags":[],"change_id":"Ibc396a191cb43111b823968c07dbecfab7ea3664","subject":"Fix docs gate","status":"ABANDONED","created":"2020-04-13 07:51:19.000000000","updated":"2020-04-14 00:38:05.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"795cf2c464a2f0afc86debbe3b39baae300ace47","_number":719405,"virtual_id_number":719405,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"date":"2020-04-13 11:25:10.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-13 11:25:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2020-04-13 14:59:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 11:25:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-13 14:59:54.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"4a8221a16e74817851c111ff41ab4232c49d559b","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 07:51:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0a7bd398792934191ab4bb9d9270c56f49bbb6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 08:41:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91c1a7bbedc3495aaf0bbd7c4f96e21e : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d287610baa44c6f94bc5cfb4350a822 : SUCCESS in 7m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/47460255a1df4b3cbde56a17655ff4ac : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d354f51ed8b406cb309605718ff753e : FAILURE in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0944639e1ec444d2a2dc150b4a361ef7 : FAILURE in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea8789e6c2014185bdf2b86581db873c : SUCCESS in 9m 03s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/271006870e2244a48483bba9332b0f15 : SUCCESS in 46m 45s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4659f20fa82843168aabee8ddcd8ab1a : SUCCESS in 41m 47s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0482952616bc41e6818410c7b0b81740 : SUCCESS in 43m 46s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/7a1c76f453054544bc06f3c638208970 : SUCCESS in 43m 57s\n\nWarning:\n  Comments left for invalid file doc/source/contributor/architecture.rst","accounts_in_message":[],"_revision_number":1},{"id":"deabaa034d41f2dd73770e7143d0bec337a4dd09","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 08:55:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5592964a356b993c5e68326af5cbd4e2451a4111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 09:48:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a5b53469bc044a48dd838d718c65bac : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a984f26771dc4882963cc3be7586ed15 : SUCCESS in 12m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/58f06adc1e8b407da51361badf9f528c : SUCCESS in 13m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a40f52c09ef944f0ae1cae89e88116f7 : FAILURE in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a553437165d46328ff8317573760a8b : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ecf51f66537483181d12f60741c5dc6 : SUCCESS in 10m 07s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/818bb6ff45dc40b09065bc0bf60a54a4 : FAILURE in 40m 13s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/07895a45d99b4c09b3a5a83ba935e57c : SUCCESS in 40m 56s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/849d1c5ada0546a7a780597de8968ea7 : SUCCESS in 44m 01s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/fd1333fe826a4e81b784f4059bbbbc10 : SUCCESS in 45m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d2378d8c2eb8a65dff05915b172432eb090e36b3","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 10:35:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7b4fd21dc4c912b59ffba0ea8eed9e8f2cc0297e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 10:48:34.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis works short-time, hope it\u0027s not a permanent solution.\n\nI\u0027ve asked on openstack-dicsuss for a wsme release, let\u0027s see what we hear.","accounts_in_message":[],"_revision_number":2},{"id":"4fe9f58ef872dd658e85f857bb64dbdab9a3d487","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 11:25:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"548bf7834b1f13380087c9fa28d22edef4d1d52e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 11:25:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65df6ad3a8be460c9890c28df7f64785 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61b5ceae558f47ab82380e22e278b90b : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d28f0fa0bbc1435f8aa320fa15f7c5ec : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd2a9ec5c4ef478a9e195eaac480f3d6 : FAILURE in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f10284b662410d96fba08ced91fc12 : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2bfb63f179f411ba41e6d9af6d855c9 : SUCCESS in 9m 31s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4dc6aa2a086c4f169aa04d1c1ba57915 : SUCCESS in 39m 02s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/91766e31ffd84658a553aeb24b59ad7e : SUCCESS in 41m 18s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/615b1352d1aa45138e332a8ffc95c3be : SUCCESS in 47m 37s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/34196e6d244146baa5715ee6c41152a5 : SUCCESS in 47m 13s","accounts_in_message":[],"_revision_number":2},{"id":"80766813f4d4a5fd9386723ad1e11d3bd1d19622","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-13 14:59:54.000000000","message":"Patch Set 2: Code-Review-1\n\nwsme 0.10.0 just got released, so this is not needed anymore","accounts_in_message":[],"_revision_number":2},{"id":"795cf2c464a2f0afc86debbe3b39baae300ace47","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-14 00:38:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2f7d95ba8c4abc982d641e429b636b31922c78c6","revisions":{"684595de5b70654cfb2be2dbb1f62a00add1a877":{"kind":"REWORK","_number":1,"created":"2020-04-13 07:51:19.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/05/719405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/05/719405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/719405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/719405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/719405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/05/719405/1"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 07:50:57.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 07:50:57.000000000","tz":-420},"subject":"Fix docs gate","message":"Fix docs gate\n\nChange-Id: Ibc396a191cb43111b823968c07dbecfab7ea3664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/684595de5b70654cfb2be2dbb1f62a00add1a877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/684595de5b70654cfb2be2dbb1f62a00add1a877"}]},"branch":"refs/heads/master"},"2f7d95ba8c4abc982d641e429b636b31922c78c6":{"kind":"REWORK","_number":2,"created":"2020-04-13 08:55:53.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/05/719405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/05/719405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/719405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/719405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/05/719405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/05/719405/2"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 07:50:57.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 08:55:31.000000000","tz":-420},"subject":"Fix docs gate","message":"Fix docs gate\n\nChange-Id: Ibc396a191cb43111b823968c07dbecfab7ea3664\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2f7d95ba8c4abc982d641e429b636b31922c78c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2f7d95ba8c4abc982d641e429b636b31922c78c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
