)]}'
{"id":"openstack%2Faodh~719417","triplet_id":"openstack%2Faodh~master~I47644f60474eb5cfc57eebaef60804900358afd9","project":"openstack/aodh","branch":"master","hashtags":[],"change_id":"I47644f60474eb5cfc57eebaef60804900358afd9","subject":"Fix py38 error","status":"MERGED","created":"2020-04-13 09:08:08.000000000","updated":"2020-04-14 09:07:09.000000000","submitted":"2020-04-14 09:05:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719417-1586855127884-11ac1cc1","meta_rev_id":"3f9985033f6393288e3084520586fcf05efb4e5b","_number":719417,"virtual_id_number":719417,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 09:05:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-04-14 08:11:37.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-04-14 08:11:37.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 09:05:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe3151bd1735801aea967cbb3cfb94ed2300030f","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 09:08:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8efb5fd4ba55e44b11e449ccded941ed8a07e76","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-13 09:27:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"23be14110af39ec0bcf7850e9389c3810bcdab09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 10:28:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5f09feddb94078932e911e06ba785e : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9090f625a03b48d49d7ec644f6a4b923 : SUCCESS in 27m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/264de05c5fb3483ea9da437c50d039fc : SUCCESS in 27m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60f38a8225e54768a5262afe78fa93d1 : SUCCESS in 25m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520058079e464adda8671ed4a551d23b : SUCCESS in 19m 34s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d78ae9b018ab443fa4ea4fafbdb2d485 : SUCCESS in 57m 14s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/2443c3b6e7f64c4eb22c71d496f6cf40 : SUCCESS in 53m 18s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f135ad6d47634a009e7adf67ab312772 : SUCCESS in 59m 00s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d1a5f926c0c0421a9fda3f6d38344970 : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":2},{"id":"570f89c1ebc7cc7f0d1c075db37dd41562abbfe0","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-14 00:38:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3cc9ad5046597268f1a43a026545d58bdd57a843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 01:30:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eb30f28de7f4bea964cb524f0c06f8f : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99a93ef569364fc1833f1a4a34dbfc77 : SUCCESS in 6m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/958d388b91cd4a2d9b2916b2bfdf7b4a : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7da4cde76a6f409fbec2fd5c2f6cce95 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d7520498ce4e9885edca89ec4b294c : FAILURE in 5m 06s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/093bc53b0bdf4b598694365943238f59 : SUCCESS in 39m 58s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/12e8cbc71e4149dab6fbbbe1a7092c40 : FAILURE in 44m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/99e2634901a04596b40789548a9896fe : SUCCESS in 46m 39s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/0a1c903a946042bd85a829ba45aceb26 : FAILURE in 44m 32s\n\nWarning:\n  Comments left for invalid file doc/source/contributor/architecture.rst","accounts_in_message":[],"_revision_number":3},{"id":"c7f5b26056fb375c38ba3f40b276318228ed1a27","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-14 05:50:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b4b7c17930496517f1ca5cb105ef9c9906a98dda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 07:06:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2ae653e2ce240aea823952b611154c6 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/178ca68d2e97493fac7ec06f235f4f1e : SUCCESS in 23m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe401e0c539f46e2a58d6fa87bdec8b5 : SUCCESS in 14m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec9c403c94674e4e812d28d2a0e20895 : SUCCESS in 12m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec00ff89953d4053980a5756e34f9723 : SUCCESS in 13m 33s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/11c6059788ce4ae0a08aa96f9e5d37ea : SUCCESS in 47m 04s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5b71dc06301f424a8ee97d268c3ce98b : SUCCESS in 1h 12m 34s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/12a9eee815434f0eb1c1e3492b16bcc4 : SUCCESS in 50m 20s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/52199a20843b4de493ffc956d6509f03 : FAILURE in 1h 14m 10s","accounts_in_message":[],"_revision_number":3},{"id":"7a41b54803b288424be72cd35555d920a4acfc6b","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-14 07:13:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"90bfabceca2f38e00e5c636655c2469047a55480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 08:03:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012a5c6fd9f34bbc9a219fec18761de5 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fe7f0f3f7e942fdb6611182867aa57a : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8daf4d49891e4057aa4165efb07ada84 : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61634a002a6a40878778e1e1db951c16 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4d5214f2ba42cb90dec1584c481eb5 : SUCCESS in 11m 30s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/75941ac7b63c458e849b18d8d6daaa8f : SUCCESS in 38m 45s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7eb3a030ca06462290087f0fc6666682 : SUCCESS in 40m 53s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e6dee87e193d44ea98c967f275faa8c4 : SUCCESS in 44m 32s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6defc07ee05b4fafbcd18208ff5f0b7e : SUCCESS in 43m 51s","accounts_in_message":[],"_revision_number":3},{"id":"1932c48dd1be2fac2aab53b243037d04938148b2","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-14 08:11:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5bff2620ac3ad2d9265c32c56fd1e0e5e1d633b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 08:11:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f7381d2c4f1c3a0a92a2f0a3350955304a588ddc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 09:05:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93ee9189300e486794fc104da904844f : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a25558a6edb04558a5be2c6f2e5591da : SUCCESS in 8m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/595326f6d82441f8932abcf88955f754 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720bfceb5e2a46509d7b02856dc9a23f : SUCCESS in 7m 48s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/958f0d5beb0443e58096d52c46bc5371 : SUCCESS in 45m 49s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5d390ecedaca477aa9977438f2f36769 : SUCCESS in 43m 24s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/593814d277bd46dea46753ea09ba1447 : SUCCESS in 48m 07s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/37610e41e6654bbbb4432c713bea090f : SUCCESS in 45m 52s","accounts_in_message":[],"_revision_number":3},{"id":"287b45ec15c16a2f7e1eb2a9f2ea709651584e61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 09:05:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3f9985033f6393288e3084520586fcf05efb4e5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 09:07:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af1d8ba4a1643fdbead76d2f3cdd13b : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9936fe5fb39c6bf38ccbb0a99a2279ae0bd8b947","revisions":{"58eb378f3f0e9e44c65d85d37f585d21a17d639c":{"kind":"REWORK","_number":1,"created":"2020-04-13 09:08:08.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/17/719417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/17/719417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/719417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/719417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/719417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/17/719417/1"}}},"commit":{"parents":[{"commit":"2f7d95ba8c4abc982d641e429b636b31922c78c6","subject":"Fix docs gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2f7d95ba8c4abc982d641e429b636b31922c78c6"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 09:00:56.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 09:00:56.000000000","tz":-420},"subject":"Fix py38 error","message":"Fix py38 error\n\nChange-Id: I47644f60474eb5cfc57eebaef60804900358afd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/58eb378f3f0e9e44c65d85d37f585d21a17d639c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/58eb378f3f0e9e44c65d85d37f585d21a17d639c"}]},"branch":"refs/heads/master"},"b7f873e25b92a20e837c3e08aec88b6dd90853b0":{"kind":"REWORK","_number":2,"created":"2020-04-13 09:27:06.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/17/719417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/17/719417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/719417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/719417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/719417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/17/719417/2"}}},"commit":{"parents":[{"commit":"2f7d95ba8c4abc982d641e429b636b31922c78c6","subject":"Fix docs gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/2f7d95ba8c4abc982d641e429b636b31922c78c6"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 09:00:56.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 09:26:20.000000000","tz":-420},"subject":"Fix py38 error","message":"Fix py38 error\n\nChange-Id: I47644f60474eb5cfc57eebaef60804900358afd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b7f873e25b92a20e837c3e08aec88b6dd90853b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b7f873e25b92a20e837c3e08aec88b6dd90853b0"}]},"branch":"refs/heads/master"},"9936fe5fb39c6bf38ccbb0a99a2279ae0bd8b947":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-14 00:38:18.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/17/719417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/17/719417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/719417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/719417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/17/719417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/17/719417/3"}}},"commit":{"parents":[{"commit":"04f507a64bf17c6113f76529d87d948e5fc57172","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/04f507a64bf17c6113f76529d87d948e5fc57172"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-13 09:00:56.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-14 00:38:18.000000000","tz":0},"subject":"Fix py38 error","message":"Fix py38 error\n\nChange-Id: I47644f60474eb5cfc57eebaef60804900358afd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9936fe5fb39c6bf38ccbb0a99a2279ae0bd8b947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9936fe5fb39c6bf38ccbb0a99a2279ae0bd8b947"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
