)]}'
{"id":"openstack%2Faodh~721173","triplet_id":"openstack%2Faodh~master~I2edf7f441cfa49c4f24bde3cccb7fbcace5cfdc7","project":"openstack/aodh","branch":"master","hashtags":[],"change_id":"I2edf7f441cfa49c4f24bde3cccb7fbcace5cfdc7","subject":"Add irrelevant-files for tempest jobs","status":"MERGED","created":"2020-04-20 06:47:37.000000000","updated":"2020-04-21 02:42:43.000000000","submitted":"2020-04-21 02:40:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721173-1587436804312-d1a6a7d1","meta_rev_id":"d7d6a1511bcfb6f5e0d5cba06b524a85b9bfa632","_number":721173,"virtual_id_number":721173,"owner":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-21 02:40:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-04-20 10:56:15.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-04-20 10:56:15.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-21 02:40:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b4ed7c865a398c6e2188255ec1745a059daca4fa","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-20 06:47:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"910735c281aa3b41e209c589b3b603c76305c279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 06:47:58.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/aodh on branch master.  The error was:\n\n  found undefined alias \u0027aodh-irrelevant-files\u0027\n  in \"openstack/aodh/.zuul.yaml@master\", line 11, column 31","accounts_in_message":[],"_revision_number":1},{"id":"8df7050947ddeabd0146b57f44a04fb79f04eee1","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-20 07:12:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14e432775e877be7c424a30dc6c752efb9684b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 08:33:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11ba3d3453d548a6852661100130bdcf : SUCCESS in 9m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/299ffacae4db45189ad5574b8d13988c : SUCCESS in 34m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a27d696e60f49668583adadaf314bb3 : SUCCESS in 31m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40ed03da51b8408297bb42a9ffbe8eb9 : SUCCESS in 32m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ddc9e94acab404da6f587ea4abaee69 : SUCCESS in 26m 43s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/30337db5eb8e426d8dde4e9a5e738bce : SUCCESS in 1h 12m 45s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/782aa29f98fc48e1a50e9711ec5f9db8 : SUCCESS in 1h 06m 33s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0642bb99701a4c65a8332ea0962a8594 : SUCCESS in 1h 19m 02s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1656b9a88968443cb2d1098496d0d54c : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3eb384faae28d05c0909d4549c3bf6161a3b1074","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-20 10:56:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"301ec888f7f7459c6184c4b20b83b282c4bf2dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 10:56:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f0ec35f98f918634743826f97eef0582df299b22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 13:08:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a7bc36ab2fe42a3957329a13b0cb05d : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b0a1ab5f15a418b854d691200a03149 : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4a62bb9d57447d290a94c9766519836 : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a796e5e9d7334d948fd267f881a93ac7 : SUCCESS in 7m 19s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c484bc8dfbaa448a9b655bac1258a379 : SUCCESS in 36m 32s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/111509b2f00545f3a9c62f1cd4cef0d0 : SUCCESS in 40m 32s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9cf148a7f945448eb2bcf9dbbd7b895b : SUCCESS in 44m 19s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/94c8b6e53a7d4a5f8d20245f232b1d5d : RETRY_LIMIT in 17m 11s","accounts_in_message":[],"_revision_number":2},{"id":"00f72cd11007e25b00cf6db0487ca7584fd4f93d","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-04-21 00:37:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"612d2d45713b6108c289961627a636fa9155f01d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 01:55:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/298877b77ed74c4b8a5ff64059f8a139 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f6ac1fb92804166bf14e9136d428fe5 : SUCCESS in 7m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85a3345bf5d64e8ca6978b073fb18785 : SUCCESS in 22m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc9fe3da44fa430a81047f26e3bc9770 : SUCCESS in 22m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3f53bd3a7294cafadad54bf8fbc2a5e : SUCCESS in 17m 09s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/9069ca93aded40aea0ed0d8d3dd5d6a3 : SUCCESS in 57m 43s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b1112bbd599c43c183b85d2a578190f5 : SUCCESS in 52m 54s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f2f5b5270c37445bb52da3001198add5 : SUCCESS in 1h 15m 12s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e7ae889b33534f2ca73c6b02d950f174 : SUCCESS in 46m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9690247711c1dce664b5e35d2f6ceb62fc0569ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 01:55:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ce1da21aa00b5e77ce0e7f9c6535ff93ccf9d98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:40:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"da964cacf15cf984f17485ca8336e5df41ed7bc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:40:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e99b2100243145199310e3989c5b124c : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f0114e0b9674dae811554f8b6cb2abc : SUCCESS in 9m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19c8dc9564a14aa287382c121434f540 : SUCCESS in 13m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ebe97daa28b46eeb8a1001ebee4ed6b : SUCCESS in 7m 12s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/18a6b2dba1234850bfc7abeff97a434a : SUCCESS in 42m 50s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/864f99b7e1094d31b17df92ab463c208 : SUCCESS in 38m 34s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e9c18b5b049e4e7f8ecc7a07f837cbe2 : SUCCESS in 38m 14s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/70efa35090dc446c822ca391fb7fbd0f : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d7d6a1511bcfb6f5e0d5cba06b524a85b9bfa632","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 02:42:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70757bb6311c4aeca57b53f27695acae : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6792853f5dc60633f3409736c81f217e88eb7b61","revisions":{"9de8e27dec7492a97dc4d3922d534d6c66f5b1ef":{"kind":"REWORK","_number":1,"created":"2020-04-20 06:47:37.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/73/721173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/73/721173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/73/721173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/73/721173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/73/721173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/73/721173/1"}}},"commit":{"parents":[{"commit":"20d03d2ebed673b00c821fc49c8888b7c969d68d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/20d03d2ebed673b00c821fc49c8888b7c969d68d"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-20 06:47:21.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-20 06:47:21.000000000","tz":-420},"subject":"Add irrelevant-files for tempest jobs","message":"Add irrelevant-files for tempest jobs\n\nChange-Id: I2edf7f441cfa49c4f24bde3cccb7fbcace5cfdc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9de8e27dec7492a97dc4d3922d534d6c66f5b1ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9de8e27dec7492a97dc4d3922d534d6c66f5b1ef"}]},"branch":"refs/heads/master"},"6792853f5dc60633f3409736c81f217e88eb7b61":{"kind":"REWORK","_number":2,"created":"2020-04-20 07:12:50.000000000","uploader":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"ref":"refs/changes/73/721173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/73/721173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/73/721173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/73/721173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/73/721173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/73/721173/2"}}},"commit":{"parents":[{"commit":"20d03d2ebed673b00c821fc49c8888b7c969d68d","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/20d03d2ebed673b00c821fc49c8888b7c969d68d"}]}],"author":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-20 06:47:21.000000000","tz":-420},"committer":{"name":"zhurong","email":"aaronzhu1121@gmail.com","date":"2020-04-20 07:12:31.000000000","tz":-420},"subject":"Add irrelevant-files for tempest jobs","message":"Add irrelevant-files for tempest jobs\n\nChange-Id: I2edf7f441cfa49c4f24bde3cccb7fbcace5cfdc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6792853f5dc60633f3409736c81f217e88eb7b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6792853f5dc60633f3409736c81f217e88eb7b61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
