)]}'
{"id":"openstack%2Faodh~721947","triplet_id":"openstack%2Faodh~master~I3b055e39544dea6173df0a7fbe9119cb150e3727","project":"openstack/aodh","branch":"master","topic":"reno-ussuri","hashtags":[],"change_id":"I3b055e39544dea6173df0a7fbe9119cb150e3727","subject":"Update master for stable/ussuri","status":"MERGED","created":"2020-04-22 13:11:05.000000000","updated":"2020-04-22 22:09:06.000000000","submitted":"2020-04-22 22:04:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721947-1587593099921-d1a6a7d1","meta_rev_id":"ee1f0fa1d0d9a379e7de468d31a769f57a8f4106","_number":721947,"virtual_id_number":721947,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-22 22:04:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-04-22 21:42:10.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-04-22 21:42:10.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 21:42:10.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-04-22 22:04:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8a04325d38c0a5be34bcba86d728d00e96a8b3a","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-22 13:11:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b2a745a1cb7d7743774ceab5c06b01821dee9c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 14:07:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae6efff2c6e4acdae4462ef3a8847b2 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/758c3a48f9c641e7a7cabec782782ad5 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbbbcfcc346341ad8104a275d09e7d62 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9af34f59482515fd5c08293e73c2e93094e6add3","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-04-22 21:42:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1cab136cc2f2c4a140ab1174625aba42c34fd07f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 21:42:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22768c57b4ced15c287032e6627adc9fd95de416","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 22:04:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47b39e13e97a4e85b974420fee703f35 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef62087ebe248a8923f1b7b18c03e39 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e6e3f643a924db2b333918196be8f94 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f1a1a75db0b23b8b8d31cd2f37f645576f792497","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 22:04:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d973c901ecd9f5e4602dbe57c5369c46eda25a21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 22:09:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b4d90de4de417fb7e38b4fca5125ad : SUCCESS in 3m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac28dc47df804ae19a92ab52e83fcb91 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"692572bd95e285c5345cf481285d7e017849238a","revisions":{"692572bd95e285c5345cf481285d7e017849238a":{"kind":"REWORK","_number":1,"created":"2020-04-22 13:11:05.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/47/721947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/47/721947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/47/721947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/47/721947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/47/721947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/47/721947/1"}}},"commit":{"parents":[{"commit":"6792853f5dc60633f3409736c81f217e88eb7b61","subject":"Add irrelevant-files for tempest jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6792853f5dc60633f3409736c81f217e88eb7b61"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 13:11:04.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-22 13:11:04.000000000","tz":0},"subject":"Update master for stable/ussuri","message":"Update master for stable/ussuri\n\nAdd file to the reno documentation build to show release notes for\nstable/ussuri.\n\nUse pbr instruction to increment the minor version number\nautomatically so that master versions are higher than the versions on\nstable/ussuri.\n\nChange-Id: I3b055e39544dea6173df0a7fbe9119cb150e3727\nSem-Ver: feature\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/692572bd95e285c5345cf481285d7e017849238a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/692572bd95e285c5345cf481285d7e017849238a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
