)]}'
{"id":"openstack%2Faodh~728072","triplet_id":"openstack%2Faodh~master~I75d52a207ec280353892354cc5fd680db5843617","project":"openstack/aodh","branch":"master","hashtags":[],"change_id":"I75d52a207ec280353892354cc5fd680db5843617","subject":"Remove translation sections from setup.cfg","status":"MERGED","created":"2020-05-14 12:53:09.000000000","updated":"2020-05-15 14:23:00.000000000","submitted":"2020-05-15 14:20:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728072-1589552421054-7d7648ba","meta_rev_id":"a7aa6cb8ba2309bc62270a404fd6fce61de25c86","_number":728072,"virtual_id_number":728072,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 14:20:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-05-15 13:22:56.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-05-15 13:22:56.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 13:22:56.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-05-15 14:20:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76d322208784cb6ea874ca782f73a0a85bc5c5b7","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2020-05-14 12:53:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bbdb5d47f21989600096342bfd09cd167d5ed3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 14:16:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe5cab86e6814c12873196fbdc04e170 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be74e7386a064181bc58d3f39399efef : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19f76ced74e148059daaba95c3cc7954 : SUCCESS in 13m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d41345353c74c43936d693a488f0921 : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b2f24d296b54a138181927e694b019c : SUCCESS in 8m 39s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/d61aa8a0581541dea2f8393cc07b7d17 : SUCCESS in 46m 25s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/29c25ef675ff4de6a98b66283df79fb5 : SUCCESS in 47m 06s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9012da0226f14f679d260823aed800bc : SUCCESS in 52m 09s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/81a8043c71c24bda856aba91e3f62714 : SUCCESS in 44m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b9efd507e2ef10f5a0cf28794f26aae2b893f70b","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-15 13:22:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b6ef96da96137820aeecaa5a00d6dfafb29f492","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 13:23:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e592031d825bfe750efdef1a3d88c737d40df0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 14:20:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ba4eb40e38c348d8985ece13620db693 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b56232190e64945bc36bf1d50a88d4a : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffed07b16cdf4642b3550309f5038533 : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/181e08c6391343c0ae40f1c789a86bab : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720adef1e12044fea1e09d2d8ca3ad58 : SUCCESS in 11m 27s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ffeb43a4b869440c9b308c1e8bd42cc6 : SUCCESS in 46m 57s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3791ea77fbc643f7a4a6249d8b09ee5f : SUCCESS in 47m 46s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/518548864fd64fc7940a60f1523ee32f : SUCCESS in 49m 00s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/935c8196b1c14b27a76b0771aa20963c : SUCCESS in 51m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4d8a6286de9543a808bc544fb657812bbd49d570","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 14:20:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7aa6cb8ba2309bc62270a404fd6fce61de25c86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 14:23:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730df68ba3934c078851af7e68fd7efc : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3127609edde37ddf669390638cb68f64f80b8262","revisions":{"3127609edde37ddf669390638cb68f64f80b8262":{"kind":"REWORK","_number":1,"created":"2020-05-14 12:53:09.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/72/728072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/72/728072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/72/728072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/72/728072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/72/728072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/72/728072/1"}}},"commit":{"parents":[{"commit":"9af9f09644af89ed770be8c93a5c06c9ec5b1ef2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9af9f09644af89ed770be8c93a5c06c9ec5b1ef2"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-05-14 12:52:24.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2020-05-14 12:52:24.000000000","tz":480},"subject":"Remove translation sections from setup.cfg","message":"Remove translation sections from setup.cfg\n\nThese translation sections are not needed anymore, Babel can\ngenerate translation files without them.\n\nChange-Id: I75d52a207ec280353892354cc5fd680db5843617\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3127609edde37ddf669390638cb68f64f80b8262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3127609edde37ddf669390638cb68f64f80b8262"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
