)]}'
{"id":"openstack%2Faodh~729115","triplet_id":"openstack%2Faodh~master~Ia07e9173b58dfd7165cc97ec70b83387fa91c3f0","project":"openstack/aodh","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ia07e9173b58dfd7165cc97ec70b83387fa91c3f0","subject":"Imported Translations from Zanata","status":"MERGED","created":"2020-05-19 06:05:43.000000000","updated":"2020-05-20 10:39:35.000000000","submitted":"2020-05-20 10:37:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729115-1589971043401-278ba6fa","meta_rev_id":"7f16133e54b85957995669a6e7568d4085123a63","_number":729115,"virtual_id_number":729115,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-20 10:37:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-05-20 10:23:21.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-05-20 10:23:21.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 10:23:21.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-05-20 10:37:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"466056ca78b5e5aa4760f3bfa6e927ebc4a6af72","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-05-19 06:05:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d3931220bce0eb6e905fe030fee30c95b4b43de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 06:26:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fa305db02d94fe3a59e440b2703e368 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b19ef68db544cf588c4a24da58c68de : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77bd3b56fa8f44f383cac6b50339bbe0 : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6e834f048fbddcec1dce220a98b63320e6fd47f6","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-05-20 10:23:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"121257a9745011e0a2fbe991f7094a593612460d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 10:23:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4f592981275bd0c0d19c7dc27773a112a592ca4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 10:37:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2825eba55734bc49b5c11576f674afe : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6284d67e054197898af7cefeaac4c8 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29ed0bd6f4c04816b86ddaa8c97c3845 : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d1a03f566f0512969275054b5c2f7795467376be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 10:37:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9d52790ba142b3cfe5937c4f9d6d96b87a157600","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 10:39:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/419e53869f3d49c89ac4675ecc2bc144 : SUCCESS in 1m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa1b86f1cfff41cd9b7bc85072c3495a : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a62aaabaace94da7e6eb8e4297a917ec7d188ac2","revisions":{"a62aaabaace94da7e6eb8e4297a917ec7d188ac2":{"kind":"REWORK","_number":1,"created":"2020-05-19 06:05:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/729115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/15/729115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/729115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/729115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/729115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/15/729115/1"}}},"commit":{"parents":[{"commit":"3127609edde37ddf669390638cb68f64f80b8262","subject":"Remove translation sections from setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3127609edde37ddf669390638cb68f64f80b8262"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-19 06:05:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-05-19 06:05:42.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ia07e9173b58dfd7165cc97ec70b83387fa91c3f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a62aaabaace94da7e6eb8e4297a917ec7d188ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a62aaabaace94da7e6eb8e4297a917ec7d188ac2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
