)]}'
{"id":"openstack%2Faodh~730940","triplet_id":"openstack%2Faodh~master~I6e47beaf0a5f2e455db5828dd2381d98aba25bd8","project":"openstack/aodh","branch":"master","topic":"cap_jsonschema","hashtags":[],"change_id":"I6e47beaf0a5f2e455db5828dd2381d98aba25bd8","subject":"Cap jsonschema 3.2.0 as the minimal version","status":"MERGED","created":"2020-05-26 19:45:47.000000000","updated":"2020-05-29 08:31:12.000000000","submitted":"2020-05-29 08:28:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730940-1590740909187-3f6755ef","meta_rev_id":"8373b23143e209f122d0edfd5d0a4d75364ca9c4","_number":730940,"virtual_id_number":730940,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-29 08:28:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-05-29 08:16:26.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-05-29 08:16:26.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 08:16:26.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-05-29 08:28:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ddfc939f0632e4ddebf98a67fe11d1deb2d67424","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-26 19:45:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4b9e3321b478800f070256517b830b0f8ac8f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 19:54:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbde5464f8f14b7cb2e9dbdfa9c01b09 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/315f0d4d1bdc45bcabdebaf56260c1f8 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf28d9137170468888fe89223ce0fc34 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65adef513d814f45903316aa6d5ec1d2 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"fb686a5f262f9169da0c28caceb7c489f3c80f09","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-05-29 08:16:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9a59c30fd6774618bafe57349c5f1a53eef9e09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 08:16:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"858647ae2ffada384972e8a5f639a9f75d280583","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 08:28:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36d082188a474b3bb9e4744e4e85553f : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50adcf0711a94ce08a9c588e97fe0a82 : SUCCESS in 8m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c6b3cd3537c48189fef736a117f7a64 : SUCCESS in 11m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08694aa70d684352a038156dc1fa471b : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":1},{"id":"18ac527a2ebfa52bba0563f93fd824d818ceb520","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 08:28:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"10210f42ef49da56aa29ec3a07064dcbad7b729a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 08:31:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c413ad45bf5b476da45beade270d41c3 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32e956517bd0dbf7436c1edd91cc49881f38ac6e","revisions":{"32e956517bd0dbf7436c1edd91cc49881f38ac6e":{"kind":"REWORK","_number":1,"created":"2020-05-26 19:45:47.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/40/730940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/40/730940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/730940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/730940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/40/730940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/40/730940/1"}}},"commit":{"parents":[{"commit":"d86bc73d3574c9d416745b9ed4d2bd74b043e133","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d86bc73d3574c9d416745b9ed4d2bd74b043e133"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 19:45:41.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 19:45:41.000000000","tz":120},"subject":"Cap jsonschema 3.2.0 as the minimal version","message":"Cap jsonschema 3.2.0 as the minimal version\n\nPrevious versions of jsonschema (\u003c3.2.0) doesn\u0027t support python 3.8 [1].\nPython 3.8 is part of the victoria supported runtimes [2] so we now force\nto use jsonschema version 3.2.0 to avoid issues, remove ambiguity and ensure\nthat everything works with python 3 in general.\n\n[1] https://github.com/Julian/jsonschema/pull/627\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: I6e47beaf0a5f2e455db5828dd2381d98aba25bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/32e956517bd0dbf7436c1edd91cc49881f38ac6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/32e956517bd0dbf7436c1edd91cc49881f38ac6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
