)]}'
{"id":"openstack%2Faodh~732755","triplet_id":"openstack%2Faodh~master~I51c6ea487e714fc61fe79a42e83a461cb0fed81e","project":"openstack/aodh","branch":"master","topic":"drop_future_imports","hashtags":[],"change_id":"I51c6ea487e714fc61fe79a42e83a461cb0fed81e","subject":"Stop to use the __future__ module.","status":"MERGED","created":"2020-06-02 17:34:16.000000000","updated":"2020-06-02 23:59:13.000000000","submitted":"2020-06-02 23:57:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732755-1591142275587-13af6716","meta_rev_id":"a67ac9e01d08acd7bdd5370a98aa6dcdf69306c4","_number":732755,"virtual_id_number":732755,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-02 23:57:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":2,"date":"2020-06-02 22:42:07.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"all":[{"value":1,"date":"2020-06-02 22:42:07.000000000","_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 22:42:07.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-06-02 23:57:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ff8d638451f20fb5e419264e0d533d4c3dd2e10","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 17:34:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a13b4731ac92700a4d4aa3a5eebbc66f5fbdc00f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 18:32:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68e9050120ca467c898091e37a40feb6 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2413fbaf99b9462da6ff2ea26e35f89d : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/732e37bc60b648a2a336e3d3fab35a5f : SUCCESS in 14m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14a3516c5b7746e89da113e966e7b4a1 : SUCCESS in 11m 24s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/043dea39ad8446e09bcc32b6daa30f77 : SUCCESS in 56m 44s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/17943d438a6245bfb219dc743e8a34e3 : SUCCESS in 26m 42s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bc2c6f53216848969e55161f496d7ec5 : SUCCESS in 45m 49s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/aadf37c9ea404f919c8cb58c21dc34cf : SUCCESS in 48m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d011819f954afbf2a9a0b62dabc9ec79f97e09e7","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-06-02 22:42:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"343dbdc01d3483c66c3d2a25cdc4dd2361ec71fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 22:42:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da260dd8b20027160ac69d92e84f3ddb0dcb05e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 23:57:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1832e9b779e46f9a5f93576c4d4ac46 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8595137172c44c8b1071e14338143eb : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63b6a87ea04944a48e1453b8e46d8b3d : SUCCESS in 10m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e635bdada2495bb06eb35ef6fd1ef4 : SUCCESS in 7m 13s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/5bdf92b6276541bd89c87f1320328e99 : SUCCESS in 1h 10m 48s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/da2d3bd73a0546c6ae38f64eb04b716b : SUCCESS in 58m 22s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/fa2156878c0342d98652c5c0a631de3d : SUCCESS in 1h 14m 23s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9106f04df62a4647b1b8fbada93db2b2 : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":1},{"id":"980a98fcdef629b0d243a438b3c6d0e104db52aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 23:57:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff8d922da8f6ef2305e066e84020284d3563dc82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 23:59:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e48d548e8e1840229ed586854f5d2ae7 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"932b1e0e6102fada52c9a1a4f7f87d5925d71bca","revisions":{"932b1e0e6102fada52c9a1a4f7f87d5925d71bca":{"kind":"REWORK","_number":1,"created":"2020-06-02 17:34:16.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/55/732755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/55/732755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/55/732755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/55/732755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/55/732755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/55/732755/1"}}},"commit":{"parents":[{"commit":"32e956517bd0dbf7436c1edd91cc49881f38ac6e","subject":"Cap jsonschema 3.2.0 as the minimal version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/32e956517bd0dbf7436c1edd91cc49881f38ac6e"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 17:34:10.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 17:34:10.000000000","tz":120},"subject":"Stop to use the __future__ module.","message":"Stop to use the __future__ module.\n\nThe __future__ module [1] was used in this context to ensure compatibility\nbetween python 2 and python 3.\n\nWe previously dropped the support of python 2.7 [2] and now we only support\npython 3 so we don\u0027t need to continue to use this module and the imports\nlisted below.\n\nImports commonly used and their related PEPs:\n- `division` is related to PEP 238 [3]\n- `print_function` is related to PEP 3105 [4]\n- `unicode_literals` is related to PEP 3112 [5]\n- `with_statement` is related to PEP 343 [6]\n- `absolute_import` is related to PEP 328 [7]\n\n[1] https://docs.python.org/3/library/__future__.html\n[2] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n[3] https://www.python.org/dev/peps/pep-0238\n[4] https://www.python.org/dev/peps/pep-3105\n[5] https://www.python.org/dev/peps/pep-3112\n[6] https://www.python.org/dev/peps/pep-0343\n[7] https://www.python.org/dev/peps/pep-0328\n\nChange-Id: I51c6ea487e714fc61fe79a42e83a461cb0fed81e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/932b1e0e6102fada52c9a1a4f7f87d5925d71bca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/932b1e0e6102fada52c9a1a4f7f87d5925d71bca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
