)]}'
{"id":"openstack%2Faodh~754015","triplet_id":"openstack%2Faodh~stable%2Fvictoria~I3240c46774614a51f0cdb38a7c07a5091b99d7f8","project":"openstack/aodh","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"I3240c46774614a51f0cdb38a7c07a5091b99d7f8","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-09-24 09:13:34.000000000","updated":"2020-09-25 06:20:57.000000000","submitted":"2020-09-25 06:17:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754015-1601014663945-0e5654ba","meta_rev_id":"b45596b5063b6dc66e3fd2f6312a797ea60d71d5","_number":754015,"virtual_id_number":754015,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 06:17:43.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-09-25 05:53:57.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-09-25 05:53:57.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 05:53:57.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-09-25 06:17:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a87749b33d0b79bad599072db3d75fd0bb62da3","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-24 09:13:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9cc4a7f917e74dc3143e9d58cf402cbb1aa62af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 09:41:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2458653824241f0b87db1bf6cb21fe0 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/871d7264778746d8a75c8ba544f4ab0b : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e06c8ebdb06042ffbc2166105291641e : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad383370fc144faabe7266306929e7c2 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ad305f36d3ee07c910a19a8e95594e618119ee61","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-09-25 05:53:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f312a0a9f44d2e50f5caeb38430bbc80744ddce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 05:54:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3870c1347fc706efeb1f084d552ced09f3ee9c1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:17:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18bfec7f92c4114bbb28d168763c7f2 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5cd3f9204074ddfab5964f17824f811 : SUCCESS in 9m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a98d2fbde4594a7ea7bd361fce3b079d : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5e9d35a04bf4e22a91f169efc2846a6 : SUCCESS in 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"11b1f9dc9b97cad86be0dec9e9bfd90879a788d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:17:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"64f7055742723c6e04a59d24e840eb30660bcf88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:20:57.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3473ba3bb1140b4bd9c3bcf25ce85e7 : FAILURE in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bea4daae1979f88ac4fbb832d9ecb88c5a714491","revisions":{"bea4daae1979f88ac4fbb832d9ecb88c5a714491":{"kind":"REWORK","_number":1,"created":"2020-09-24 09:13:34.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/15/754015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/15/754015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/754015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/754015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/15/754015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/15/754015/1"}}},"commit":{"parents":[{"commit":"8635b67777e3aa0c90eccdbcb5753e33a1750aeb","subject":"Use SETUPTOOLS_USE_DISTUTILS\u003dstdlib for global pip installs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8635b67777e3aa0c90eccdbcb5753e33a1750aeb"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-24 09:13:30.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-24 09:13:30.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I3240c46774614a51f0cdb38a7c07a5091b99d7f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bea4daae1979f88ac4fbb832d9ecb88c5a714491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/bea4daae1979f88ac4fbb832d9ecb88c5a714491"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
