)]}'
{"id":"openstack%2Faodh~754023","triplet_id":"openstack%2Faodh~master~I3e587330d9130faa3fa12a2d9460dad171691d59","project":"openstack/aodh","branch":"master","topic":"add-wallaby-python-jobtemplates","hashtags":[],"change_id":"I3e587330d9130faa3fa12a2d9460dad171691d59","subject":"Add Python3 wallaby unit tests","status":"MERGED","created":"2020-09-24 09:13:49.000000000","updated":"2020-09-25 07:05:48.000000000","submitted":"2020-09-25 06:56:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754023-1601016968762-b51fe808","meta_rev_id":"47cc95354f1a82c4e162cc394037ca2f683ae8da","_number":754023,"virtual_id_number":754023,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 06:56:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":2,"date":"2020-09-25 05:54:18.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":1,"date":"2020-09-25 05:54:18.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-25 05:54:18.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-09-25 06:56:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"488f548ff12cd62fa055a24fb9980a686cb4fb65","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-24 09:13:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a2567a2e333fbef67dee6bf54e5ac78e7b56018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 10:55:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08a09076994f4962901d0405688f1827 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd278e068622437daaf05d6c8f41afa0 : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdf5d47ae8eb440597609e2fc8bb58b5 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5114a01c53f413b9d287c8554c416a4 : SUCCESS in 7m 03s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3bd44c8d19044c38b22b3be35f6f8515 : SUCCESS in 56m 28s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b472f80b55c94122b56b107700ccac61 : SUCCESS in 45m 31s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9c62cbe2643943e8b599905298024834 : SUCCESS in 54m 56s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/fb1bf31108e54051a26b32c459bbf3e5 : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":1},{"id":"77185ac8b2cb0318b0a6d57b73cdbeec189528c0","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-09-25 05:54:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0d6aeafd08a6f18f4d7eb7d7b822e17b966a03e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 05:54:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95d5ff8f83efc0b57951e87869b4ade0baf4b4c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:56:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9803d747864f21bf319347013dc65d : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd5df9a25f2245f182bea55f6d11e910 : SUCCESS in 8m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fed29cbaecb4569b84c6a1bc78fff90 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/347625821d6345a7b264187a3ee4a864 : SUCCESS in 7m 59s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ed9a15e757594f84b361e55d47df17f6 : SUCCESS in 46m 53s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/47aa0cd0522649ae953c85a209c94e9d : SUCCESS in 49m 17s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b7ad95067af544e5a551eaaa20cdc13b : SUCCESS in 55m 07s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d1e984024d4e41e5a033603b291e198c : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1f19b0b5920309b0a7a12c813558469ed99d0c5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 06:56:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"26401044854aa8b39bd501771e4bde039e7bb9ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 07:05:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac1950a9b4ca4e5ebd58357f53ac4430 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"267f07ed6f790a8f818f155d4fa7b87596412504","revisions":{"267f07ed6f790a8f818f155d4fa7b87596412504":{"kind":"REWORK","_number":1,"created":"2020-09-24 09:13:49.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/23/754023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/23/754023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/23/754023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/23/754023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/23/754023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/23/754023/1"}}},"commit":{"parents":[{"commit":"3399daec53642fd3caa4746aa4805f0c4c19e709","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3399daec53642fd3caa4746aa4805f0c4c19e709"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-24 09:13:44.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-24 09:13:44.000000000","tz":0},"subject":"Add Python3 wallaby unit tests","message":"Add Python3 wallaby unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for wallaby.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I3e587330d9130faa3fa12a2d9460dad171691d59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/267f07ed6f790a8f818f155d4fa7b87596412504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/267f07ed6f790a8f818f155d4fa7b87596412504"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
