)]}'
{"id":"openstack%2Faodh~764978","triplet_id":"openstack%2Faodh~master~I030a5c6a94c5a4be59051d1b8828ab41281465f0","project":"openstack/aodh","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I030a5c6a94c5a4be59051d1b8828ab41281465f0","subject":"Add common personas to base policies","status":"MERGED","created":"2020-12-01 16:12:30.000000000","updated":"2021-01-12 11:35:59.000000000","submitted":"2021-01-12 11:34:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764978-secure-rbac","meta_rev_id":"d4772d5cd204b3aa1bfa6f13c7f50ebb681f30c5","_number":764978,"virtual_id_number":764978,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-12 11:34:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2020-12-03 10:36:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2020-12-11 08:36:47.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2020-12-11 08:36:47.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 17:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 10:36:44.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-12-11 08:36:47.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"87212f5ccb6e88042cb3cd4d1b4c256107f0a488","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-01 16:12:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fe83acc01e80ebe155f914f55d570298fbb6244","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-01 16:15:10.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"7572cd9ca4c6ad55d0ad9021646d402e55984d0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 17:28:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1e434d7a4c4914a16cfe187f72a2ca : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90f2e54ddc6f4359a4fc39aa7396808d : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0763dfa9084944c3a98113ac5019b73b : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0261c84c7ba45258f8d04f9bd76e298 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89364549db54cccab19e6c97fc7c02c : SUCCESS in 7m 25s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/27da53ff117e46be9ab6e2d8946d2ec3 : SUCCESS in 53m 43s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e01b527db513401e95d034dc5580ee17 : SUCCESS in 42m 23s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/de26e6f0509b43d595db578fa2047344 : SUCCESS in 56m 09s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e049c364edec420fa945aaca709d9da8 : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4165a0bdf8c35be8b3485ca273734c9121a351a2","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-12-03 10:36:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"588290a1d74f8664ce3194aa9d302c2c1fc2f84a","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-12-11 08:36:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d00c092018e496b4fe06e5fdef31715ce6b4c045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 08:36:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e12836cc3ec672050d4d73da54e0d1a86336200","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 09:21:16.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b394e055ed5941ef922cd211750cb88c : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a41ec431e7c14cfe9559c4bf07e746f8 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9040f4c5cc244daa80152630a392746e : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ce075a92ac43119057d60f800be301 : SUCCESS in 8m 58s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/aa988b3c122145b1bec18d3dc5fa728b : FAILURE in 28m 01s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1ad57cf5558d4f6e98884dc6fcda7912 : FAILURE in 29m 15s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9585a75c92694a1384b4675156581347 : FAILURE in 37m 26s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/bba3725a88954cfa8ac7f06b4bcb7237 : FAILURE in 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8dc1a0f06a08b9bbb9f2959d23f33557ddd5024c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 09:21:55.000000000","message":"Patch Set 1:\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b394e055ed5941ef922cd211750cb88c : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a41ec431e7c14cfe9559c4bf07e746f8 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9040f4c5cc244daa80152630a392746e : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ce075a92ac43119057d60f800be301 : SUCCESS in 8m 58s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/aa988b3c122145b1bec18d3dc5fa728b : FAILURE in 28m 01s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1ad57cf5558d4f6e98884dc6fcda7912 : FAILURE in 29m 15s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9585a75c92694a1384b4675156581347 : FAILURE in 37m 26s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/bba3725a88954cfa8ac7f06b4bcb7237 : FAILURE in 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f0352dffe18551b7743f6ffee40843e036d737f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 09:22:46.000000000","message":"Patch Set 1:\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b394e055ed5941ef922cd211750cb88c : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a41ec431e7c14cfe9559c4bf07e746f8 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9040f4c5cc244daa80152630a392746e : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ce075a92ac43119057d60f800be301 : SUCCESS in 8m 58s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/aa988b3c122145b1bec18d3dc5fa728b : FAILURE in 28m 01s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1ad57cf5558d4f6e98884dc6fcda7912 : FAILURE in 29m 15s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9585a75c92694a1384b4675156581347 : FAILURE in 37m 26s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/bba3725a88954cfa8ac7f06b4bcb7237 : FAILURE in 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"bebba109fcd7a7b8b29fc9f7e688df8224c24f40","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-12-11 09:28:11.000000000","message":"Patch Set 1:\n\nFails due to https://bugs.launchpad.net/bugs/1907667","accounts_in_message":[],"_revision_number":1},{"id":"bd7256a1ba8e22e5bd08ccdaf6b7b3c052cfc254","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-12 08:48:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e321ed5e8c0511557f2503b80257320450b3198d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 10:09:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58038feab2644252abee250b792065e8 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/679fe718f71c49d2b888831d2bb0f5fb : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e981f3d3e37e4442b89474c01261fed3 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3de68d6f69445d5bc3d942bceb18081 : SUCCESS in 16m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb8604463764b2daefc9391644a6d89 : SUCCESS in 7m 31s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f72e388b0d8f47349fed874d65ccf1ca : SUCCESS in 41m 38s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7c8a77bfd3b54ddd91a31009a2dc3c08 : SUCCESS in 1h 00m 03s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b055c20397c947cab749d78e09e51c18 : SUCCESS in 51m 19s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/143591c339a449a9abc4db5a552d1821 : SUCCESS in 1h 05m 17s","accounts_in_message":[],"_revision_number":1},{"id":"289c88e0b46d2c5a72832c1c35f73cec4fdf39dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 10:10:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"535b58fe52656a0a4fb8dba5d8ccba5f99be61ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 11:34:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0f99a204134d928938f8d3d8ff7619 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b976fb08bd1e45e388479b09eefa61f7 : SUCCESS in 11m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5d4715ff61b47ccb4cef53579c1a340 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19b7502aaaa54da7ba94eed5a2851f46 : SUCCESS in 7m 41s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/7ed29a937f54459aa017da5ecff61919 : SUCCESS in 1h 17m 30s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7f3ca418f69445af82f934593eef08fc : SUCCESS in 49m 01s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/99bdba2ce0414b4f88c538451fe4bd20 : SUCCESS in 57m 44s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/da76c6ee9e804004876f39c3c908145b : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":1},{"id":"3f06e9b5fdca6104553396cbc134cc0ae59de88e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 11:34:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d4772d5cd204b3aa1bfa6f13c7f50ebb681f30c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 11:35:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ada1084b904b398dcd82c5f8170d11 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5df5505c8d29ac35531680219609bd704fdbc847","revisions":{"5df5505c8d29ac35531680219609bd704fdbc847":{"kind":"REWORK","_number":1,"created":"2020-12-01 16:12:30.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/78/764978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/78/764978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/764978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/764978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/78/764978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/78/764978/1"}}},"commit":{"parents":[{"commit":"b99761d38f82e198f60ce54613698db52cc6d1c8","subject":"Update requirements to implement secure RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b99761d38f82e198f60ce54613698db52cc6d1c8"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-01 15:45:53.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-01 15:45:53.000000000","tz":0},"subject":"Add common personas to base policies","message":"Add common personas to base policies\n\nThis commit adds some common personas, or check strings, to the\npolicies.py file. We can use these later to implement more secure policy\ndefaults that will be consistent with changes across OpenStack and will\nprovide a better, more secure authorization experience for operators and\nusers.\n\nChange-Id: I030a5c6a94c5a4be59051d1b8828ab41281465f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5df5505c8d29ac35531680219609bd704fdbc847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5df5505c8d29ac35531680219609bd704fdbc847"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
