)]}'
{"id":"openstack%2Faodh~764979","triplet_id":"openstack%2Faodh~master~If17dbfec10302c9c57d814cb406faee1a6efc59a","project":"openstack/aodh","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"If17dbfec10302c9c57d814cb406faee1a6efc59a","subject":"Implement secure RBAC for alarms and quota policies","status":"MERGED","created":"2020-12-01 16:12:30.000000000","updated":"2021-01-18 17:25:10.000000000","submitted":"2021-01-18 17:23:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764979-secure-rbac","meta_rev_id":"c8743285c57cd9e29231396b5b501807f8770823","_number":764979,"virtual_id_number":764979,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 17:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2020-12-03 10:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2020-12-11 08:36:58.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2021-01-18 15:08:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2020-12-11 08:36:58.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 17:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-03 10:39:16.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-12-11 08:36:58.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"862410e5edeca4bf9fe99504e121853c82496d82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-01 16:12:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52b178a61f966e926e21ab20bdf656796ac9d6b8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2020-12-01 16:15:32.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":1},{"id":"c8c06d5e3330c65254fa7e3b26739eff6f8b85bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 17:27:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d87343952a4b4dd98368e47478c615bc : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/296cbaa244e943a2b3cda754479f8187 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0189c1ab32d462c977b71341dab4ab8 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1696aea866a410c90a10180d2dd37fd : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c767663912fa4433ae02b2c34ad7ce68 : SUCCESS in 7m 03s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f1d222df2b284481a3b01876f3797f19 : SUCCESS in 52m 34s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7ff7425ce95c407cacf66e733c6c3ee1 : SUCCESS in 53m 54s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/7b379c1f340142a386aaa2414df5b2ab : SUCCESS in 53m 51s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c9f0168d232241b5b152f6bf16c99009 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7ac4495db33f5895300a1d5b709bc6b871d2f585","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-12-03 10:39:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c8f96c4933dfb58c1e2d97920f59b3c18e6b9fe7","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-12-11 08:36:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"234e8198bc721cc0d35955084e6e33f0b33d9c42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 08:37:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d05575561888915e497363c573d3f66d7d004db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 09:24:15.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"fb9968b552b1382a8de46ed784f50e3dd450e47f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-11 09:25:06.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"a3efb181f5deb8f5ee11dd05f0a5b078d2b68d33","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-12 13:47:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e4534696438f27228b6145770f8c3a079a54945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 14:45:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a369d24e5264536bbd2ba95df0b61a7 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9353348f58f54888862f3032ba424cac : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ff0548d28cc408c9a8a4ff3126343cf : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8aa97dc3e70443618759a40fa71e20f0 : SUCCESS in 23m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa0d1a259f448ed8705daeb0bff98b4 : SUCCESS in 8m 49s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/058a9f3c556540b59b239d74780eaba6 : SUCCESS in 46m 49s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/c91b68621d07428780594e5465d2e71e : SUCCESS in 51m 07s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/693cfc567da34309b16e9c61a5a8bde7 : SUCCESS in 49m 01s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/1772748a99db4c46a13d7f71ec002542 : SUCCESS in 49m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9a001b6b605e541f22557c1c75fcd4a16421afb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 14:45:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b23bedb172e4e0207a9f46dfb2c0b056dffe94e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 15:50:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c25d9f7ea74d8aaec77b708cb59aa8 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce7946ad04b24f3d862d626bcc855825 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cb984a41e7b4f378cbbd5ee8116e685 : SUCCESS in 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81573b6f90484eafa36d22831bc5748d : SUCCESS in 10m 05s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/67705d4dc7ec4a9c941c094eccd4f16c : SUCCESS in 1h 01m 34s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/0c880784569045cd888439b6be8a4f85 : SUCCESS in 38m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d650a3da3c0745adba8a1e98e4e003e8 : FAILURE in 30m 35s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/58600f0d7b2447e097e46bc5bd08b1aa : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8e6a4851bea8242cab04f574d21de76868e3020e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-13 16:47:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"335229170ce894c73f3d032a3555d58352c1860a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 18:27:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576277c2cd9446f6966ae624b50a0fb1 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfc651e3d2b54de98448ea30b4a8c19d : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d01a3124dbd1474db18a582b73664c33 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8eef87045a8e4ddb9553ee7164d56c49 : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df884ba2cfe4d95bb92b0f0bc63ca08 : SUCCESS in 10m 34s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/b910d6228f564767bb0436bb54f780a4 : SUCCESS in 1h 29m 35s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/100183b3e7bf42f88eb35033d5c788b7 : SUCCESS in 43m 48s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/6fee8e5a50ac4dc689941a231740f0d2 : SUCCESS in 1h 32m 55s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6c7b2d3d3346489683fdc59074b829b3 : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1998293a110a6779ab58d2579b385427316ff2bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 18:27:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47e8a033b2d6c842bb27a8771d3968a13cbb52c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:24:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7593c47e453748cf97ebfcc159b0e81c : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35f626a914bf42cfbd50c5ca9c0b4381 : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/284c3eac992a429ca02dd0dd51cdda07 : SUCCESS in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26826cd92fc54750a43a41d6b5e19b78 : SUCCESS in 9m 37s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a8834e4b62614f3d96646e04d6ddf4a4 : SUCCESS in 53m 05s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/a7f33c22a1e6410a90e14ad8c7f60c5c : SUCCESS in 50m 25s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/918cf4053ff84901ac8f1f9ad9797022 : POST_FAILURE in 47m 05s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e5af2438d34d441d98a0ec47dd2477b2 : POST_FAILURE in 43m 54s","accounts_in_message":[],"_revision_number":1},{"id":"dff116d29f9407ebd274505e3637e674ac04c6ff","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-15 08:35:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9f504de22dec5921661ac5ce084cd22c75c89b01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 09:51:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d7f8cd07a3c45aea2352373fc62ca05 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6776eeee03ab4383a2ea2644a9a74d53 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8500b42ceb54df58f3063682093282d : SUCCESS in 14m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99775f497ec14d368645b5d52c1ab610 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aca5434010cd4f11898cb92c09a05090 : SUCCESS in 9m 03s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6b1a55ee8cd34e379c8d9e4cc71a2eae : SUCCESS in 1h 00m 30s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/be11bf000a664d48a92afce0d604e944 : SUCCESS in 1h 12m 40s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8d613c783760450e8b660d2db139fcd5 : SUCCESS in 51m 00s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/11cbf75987b44414a4e28d2fb4caf552 : SUCCESS in 53m 02s","accounts_in_message":[],"_revision_number":1},{"id":"bffaee0d02f952b4dc9d82b3251c9bba514ad426","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 09:52:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f799b3e0bc8f70cef7ba54e269af789a218485f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 10:44:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a9ffa4a51a04526ab7a703a00071fde : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d673b8c4d304a5da03db7e9b1e8a8ad : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d4184b7cec2479db9c6d4864955e56f : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ebc8c60fcd44dab95f70f5e1944b5c : SUCCESS in 7m 53s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/43192c3495db4fa6b8052acf5310ed0f : SUCCESS in 49m 51s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/70dad2765d344eaa85abf601f9fa3d4d : SUCCESS in 46m 50s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1b8e0e2e8a274dbea2e3d9ae52f590fc : POST_FAILURE in 39m 43s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d5f84b8b678445d2a6f5c8fb47639f89 : POST_FAILURE in 48m 15s","accounts_in_message":[],"_revision_number":1},{"id":"af9937ffca3ccb7d4367b6c464d040453c3151c2","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-18 15:08:05.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4ae1f92c2ba29f90006c3df24bf171759e587c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:12:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b1440219c25452f81dde108254a872e : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf44f085948c4d7fb2ec596bcc763c06 : SUCCESS in 9m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bf13e9ae6624be0a801754aaae2f9a0 : SUCCESS in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9de1e6d29e964e619d57ec47fc0b5bbc : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdee04074ac3477284d262b207b2f3ba : SUCCESS in 7m 53s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c7dc0c499f7d4ad5862c96235b7ce041 : SUCCESS in 44m 28s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d89f9541929548e29aa2244cb96f4b51 : SUCCESS in 50m 42s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a6735b519e67446e96cd96d433e709b2 : SUCCESS in 45m 42s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9a2c18874af642e68dcc6804a54cdec5 : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":1},{"id":"777512e763a24080f3fa68d36bce60d60ab8f3dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:12:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51c49a5879f55891d518ff670da90be2f714111f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:23:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ab35cc9ebd24bd3b45f42e7a2a2233c : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71e7a19d614d4e51aacb7a78fcb4556a : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/873e0ca83d384bcca3e8337c13a194dd : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc9efc5741ba4726a117232e35403b29 : SUCCESS in 10m 26s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3d1afaf1c4aa49f8bcf2d2dd48ade5c1 : SUCCESS in 50m 15s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/eccf7926221b42fa8a5c50be8e4598c3 : SUCCESS in 43m 03s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0b566b955daa4cc89cc379d62fc968d0 : SUCCESS in 1h 02m 18s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/f3ec3d38fa7149d19ea2998440193fcd : SUCCESS in 52m 13s","accounts_in_message":[],"_revision_number":1},{"id":"27d41f6a083ec4cb762e6c841ee52c9e13330fc9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:23:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c8743285c57cd9e29231396b5b501807f8770823","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 17:25:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba0959a1acb432881f57806e378f7c5 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7779da4916ce4c485ba9c8db04ca4dfcacb7ba2d","revisions":{"7779da4916ce4c485ba9c8db04ca4dfcacb7ba2d":{"kind":"REWORK","_number":1,"created":"2020-12-01 16:12:30.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/79/764979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/79/764979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/764979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/764979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/79/764979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/79/764979/1"}}},"commit":{"parents":[{"commit":"5df5505c8d29ac35531680219609bd704fdbc847","subject":"Add common personas to base policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5df5505c8d29ac35531680219609bd704fdbc847"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-01 16:03:39.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2020-12-01 16:08:16.000000000","tz":0},"subject":"Implement secure RBAC for alarms and quota policies","message":"Implement secure RBAC for alarms and quota policies\n\nThis commit updates the default policies to include supports for\nenhanced token scope checking and default roles, including a read-only\nrole for GET APIs.\n\nThis is part of a broader change across OpenStack projects to\nconsolidate common personas and provide a more secure and consistent\nauthorization experience for end users and operators.\n\nChange-Id: If17dbfec10302c9c57d814cb406faee1a6efc59a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7779da4916ce4c485ba9c8db04ca4dfcacb7ba2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7779da4916ce4c485ba9c8db04ca4dfcacb7ba2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
