)]}'
{"id":"openstack%2Faodh~768499","triplet_id":"openstack%2Faodh~master~Ia0fa4ba1fecf2740a291c5a65b906b089c95404d","project":"openstack/aodh","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"Ia0fa4ba1fecf2740a291c5a65b906b089c95404d","subject":"[goal] Deprecate the JSON formatted policy file","status":"MERGED","created":"2020-12-25 03:52:05.000000000","updated":"2021-01-13 20:14:29.000000000","submitted":"2021-01-13 20:12:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768499-policy-json-to-yaml","meta_rev_id":"68f2747a6c7df9f3806a5d3143778e1dd1894b7a","_number":768499,"virtual_id_number":768499,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 20:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"recommended":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"all":[{"value":1,"date":"2020-12-27 10:46:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":2,"date":"2020-12-26 15:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":1,"date":"2021-01-13 07:05:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-25 04:37:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-26 15:24:36.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2020-12-26 15:27:27.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-12-27 10:46:53.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"}],"messages":[{"id":"c568fe478679222c0d00c37bd19e1e9ea3f1fd5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-25 03:52:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f727b82f78743f1857273da33a4d2ff7fe24cc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 04:37:15.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/984ee484328d4798bea28b40f7bc9798 : SUCCESS in 1m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27fa853c1c8f431ebbff73239da99e79 : FAILURE in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36e77f74c7c3451fa243bf7af604b41d : FAILURE in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c5a053c809343c38f16e24d10a4f263 : FAILURE in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78dbd4c7b44d4b7eb8b72881b1940250 : FAILURE in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9676142eb3b04bb696583632c32566f9 : FAILURE in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f016e3969e3c440ca3f42d544670a0a1 : SUCCESS in 5m 10s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4328a61034654b508f1c36b7d018ed85 : FAILURE in 41m 58s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e20fe48536dc4a0995795647c1a29f70 : FAILURE in 33m 27s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/20eebd1f4b0149f2b62d123374555519 : FAILURE in 40m 24s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/5b8bfe8be5034e95bbc52bac36bbb266 : FAILURE in 36m 40s","accounts_in_message":[],"_revision_number":1},{"id":"62d51074e0942c7f732ac51fa1bbcc99697c1d7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-25 15:40:34.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0d435eea3a9bbbdaba5eee1396cda1dd2399a8b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 16:17:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/35cf3aaa872c4c069bd03e36c53789db : SUCCESS in 1m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbedae7c9d014360a3dd20e85bd7c0ac : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4879f11c121a4e99a7d9023be2051949 : FAILURE in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b203ce9cf084712adf8f68108368b27 : FAILURE in 9m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/491a9792f09d4e738d757ea4d5df0ef7 : FAILURE in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e294bd69c454a488552b4f83cec4305 : FAILURE in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c80116e765fe441f8fce2f3abb38b773 : SUCCESS in 4m 43s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/90644da266f34378adefe7ec2786bb96 : FAILURE in 28m 20s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3479c7f9d5e9471aaaac6653de99a043 : FAILURE in 30m 41s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/3f392eb1c59e4c7581bb3a4707a3f400 : FAILURE in 25m 03s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/48caa8a7ed8746d6ba43b7ade7897d96 : FAILURE in 31m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5ac4cf086e39e88ec8b04ee23a9b3d554069bda9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-25 16:41:56.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5d8df202b4f8439436c3b2a42d9ffb2c9fe448f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 17:20:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9053e62b94164f0ebc36f2fa292e57d6 : SUCCESS in 1m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/337b3eee983543cb892aa72334c28f43 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b74861cfcb7c4337883cbab754d132d8 : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87a92a713c8d4893b6c9e56187d3b71b : SUCCESS in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5ebeccf0f524d07a4a9149b0657b45b : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/317827a34a5b450590c27f95e49170e5 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46a27a4a97ea4146867f6604be601b83 : SUCCESS in 5m 09s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/e8fafe6effbc4efda3da45d728d684b7 : FAILURE in 33m 28s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7c5b3b26f2d34666ab61b5e63be876ad : FAILURE in 30m 58s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2663f1e399024b619dc9be8d9c768c96 : FAILURE in 24m 46s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/ce0371ec75b84c7fa38a5fa9e6c6503d : FAILURE in 32m 07s","accounts_in_message":[],"_revision_number":3},{"id":"dbe8e9989ad57d249efe024a131e6c5406fdb989","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-25 18:26:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"45ccf8c147678d02cc3eb6d7ec1e8ee7b23745ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 19:03:03.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c8b1198338a041f5847b25fa5208c2dc : SUCCESS in 1m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a562edb96447f18554887ae771eacd : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24acc12fdc03479cb6351865bb14877e : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08488c2025fd4cf0a594e1ea11060f2a : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e1195dd9580495aab00f83adc2ac56e : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6212f6a03444e27a776765db4f7b4d2 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5065e1b6f564264a81d3e20634e8338 : SUCCESS in 4m 47s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6851ccceae0d4a798d3ee99a9ad76048 : FAILURE in 24m 13s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5b1f7ebad42f4583ac23f3fae8ef4258 : FAILURE in 31m 07s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1935592d3e4340d7a16af8c858ab53a6 : FAILURE in 24m 30s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d48b7e6c4b784fbcb20062c7b934c3fd : FAILURE in 21m 46s","accounts_in_message":[],"_revision_number":3},{"id":"b9701ceff1926290a461fa8ed2e3d81b498e031c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-25 22:22:44.000000000","message":"Patch Set 3:\n\njobs are failing as this patch need oslo_policy 3.6.0 and gnocchi is capped with 3.4.0 https://github.com/gnocchixyz/gnocchi/blob/e19fda590c7f7f07f1df0ba93177df07d9802300/setup.cfg#L33\n\nwe need to wait until gnocchi start using oslo_policy 3.6.0","accounts_in_message":[],"_revision_number":3},{"id":"81eaf41598c115f419b5d031b04b8cf2c9912323","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-12-26 15:24:36.000000000","message":"Patch Set 3:\n\nRight, I\u0027ve opened a bug and a PR for this\nhttps://github.com/gnocchixyz/gnocchi/issues/1095\n\nand https://github.com/gnocchixyz/gnocchi/pull/1097\nUnfortunately, the CI is unstable (it fails the same way on a docs patch)","accounts_in_message":[],"_revision_number":3},{"id":"d4dbe66385f2c7e99fae8951897abb2366743eed","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-12-26 15:27:27.000000000","message":"Patch Set 3: Code-Review+2\n\nThank you for your patch, it looks fine to me.\nNext step would be to fix gnocchi.","accounts_in_message":[],"_revision_number":3},{"id":"3828233c09265dacbe64baa59ac9fcea6e119119","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-26 20:10:40.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e Right, I\u0027ve opened a bug and a PR for this\n\u003e https://github.com/gnocchixyz/gnocchi/issues/1095\n\u003e \n\u003e and https://github.com/gnocchixyz/gnocchi/pull/1097\n\u003e Unfortunately, the CI is unstable (it fails the same way on a docs patch)\n\nthanks Matthias.","accounts_in_message":[],"_revision_number":3},{"id":"6ee750cf1df259712c4c03ebb2dedaf8cd685a1c","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-12-27 10:46:53.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"979b9a60a6ab0ed0548999bc1bcfa94d5d8fe231","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-12 08:47:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"db0ecb006e2b3d9e6b1c33cecc647bfa7ecb5fbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 10:13:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/40d4cd5876f04ffe8f193ed118f9e86f : SUCCESS in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57ab90751c214a07852fae0e70b6e528 : SUCCESS in 8m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24da1aec869d4844a4e8562be9398582 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fadf06b1a18045f88fce554025d9f7eb : SUCCESS in 9m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d15aa8b229b84ea892d1bb7ba9534c32 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea32526353c4b25970f2e0224c18693 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d73bae548e084619a9b65dee2ae4139c : SUCCESS in 4m 50s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/3417a0ba3d384809aa441ab385289e82 : SUCCESS in 47m 19s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3797d8111f6348fbb50b2459319f6309 : SUCCESS in 1h 05m 06s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/44ae8bd059904063bc38889dc8aff724 : SUCCESS in 1h 13m 01s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/29c5fea3b6b240928663d41e735dc4b4 : POST_FAILURE in 43m 06s","accounts_in_message":[],"_revision_number":3},{"id":"ca1f618cd369f54f67b19b7648d23ee6e4c77e50","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-12 13:47:49.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"dbbab50ed711af1abc29a75e1bfb727cdac7aa38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 15:33:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/882668f062504ecabfb973f8f91164fc : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b2de66e514478eb36c8e45026e9e87 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9999b578421b4a38aee7d7f5357be030 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3b7f0c255334dc3b0da20f4a0b0c01b : SUCCESS in 8m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae39113ae4054c1aa1d9a14ca99e5a8b : SUCCESS in 37m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7536753bb6f4430c8441d37593b872e4 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b135456505dc4a38881e4ae1a914d77e : SUCCESS in 11m 49s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/209799dafe284fb1a413a6b5afebf0a3 : SUCCESS in 51m 20s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/5842e4995ab64b0a836cc529afc8f664 : SUCCESS in 40m 38s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/17c5b49011b44a0db6341d1172989be2 : SUCCESS in 54m 08s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/e00c76b29a1f421daccc096fb8dc77df : SUCCESS in 1h 36m 44s","accounts_in_message":[],"_revision_number":3},{"id":"919d3fb314e6cd084cf55b1e3302eca25766d483","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-13 07:05:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1d7ac9e5a91b3ec02a849c2f25f3b693a5f29a61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 07:05:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e67b2917c16f6b3658f31740337f086b8b5c52b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 08:10:33.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/389e46c6fc2d4faba0f8148130f434a1 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ff50f0c2004167914c00a69c5725c7 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7402f2718ed4d6cabed9b8311cae656 : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4377c95f80747ba9d20005b90989bd7 : SUCCESS in 10m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/084faa7d4e5d428b9f7728485b1b0279 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/565f5a6f3f974d1a95ff5c476d71ce0a : SUCCESS in 5m 37s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/31314a613ce64bfc981579087006ab10 : SUCCESS in 50m 45s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d4936425d0e548e898900fd89c6fedad : SUCCESS in 50m 48s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/0fef0a76932f4b6695a26234d0a884a7 : SUCCESS in 55m 03s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/dcb4667fc0f2493baa8052182def2431 : POST_FAILURE in 47m 03s","accounts_in_message":[],"_revision_number":3},{"id":"7d40be27bd086f8a6eb9831a95659b74941be36c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-01-13 17:38:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4eedbb553cf44ecfce3a08232ed6ebcf11e36679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:12:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0b0d79bb1602436cadf91b32b061e995 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53a42b36c55d43c9b770a2ecfa3c9c26 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0438b04e95794c9a9729d17f8ee69e84 : SUCCESS in 12m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/432cc3ac56f646688bb51536b7716aba : SUCCESS in 16m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ab7734bd2474a59a6e7710349c9e582 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0bf7fddea44dac9b8cc17705af7016 : SUCCESS in 21m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e16892f8758c4b73b4bb871e43d6f980 : SUCCESS in 8m 33s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/368ce551f8e049afbbfc7f8dc8494e2a : SUCCESS in 51m 41s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/3bb05ade69a04badaf75aa2abfe6cda6 : SUCCESS in 1h 18m 39s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cfd767c82872438898874ae7a117ea96 : SUCCESS in 1h 02m 37s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9e4fbfbca3c44e289b059dea0c87b440 : SUCCESS in 53m 46s","accounts_in_message":[],"_revision_number":3},{"id":"734fe745c09904d755fa091baf096e956ded4057","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 19:12:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3ae8a91aa8c98bc86ca113fb21f66e0bd71d51d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:12:42.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/738d50da98c3445ebf8f0623bd9a4cd8 : SUCCESS in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/250da9fd84774a9ea8b52fd751b3a7af : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/106e62b6c8b2445bbd94f79342804d21 : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9af8fb6debc54d1d912f8e9644df2225 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3027a58cba448668dffe26f4be09805 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9eb68597da94c46ae46cd0174b3c042 : SUCCESS in 7m 02s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c49dcdc37ad04ada8adcd22b0a42d8e2 : SUCCESS in 52m 37s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/29ac8c55c5274f93bf9ab1a9fdd61dd9 : SUCCESS in 47m 49s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/2f349225088445cd97c9a4fd2db93026 : SUCCESS in 51m 20s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3b4d4887896a496893f424f5539c9bc5 : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":3},{"id":"948b171c3271c0cb916d909be6430bd5145cbd71","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:12:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"68f2747a6c7df9f3806a5d3143778e1dd1894b7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:14:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/457cb29691bf42df937d8adea7644985 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49cbe5d11a4c417d8d4fdcf83e45619e : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5e7a0519cfc246dd2b5ca9a9749428cf0e5217e8","revisions":{"552a20572f19ed0f60de863700583c722d7673a0":{"kind":"REWORK","_number":1,"created":"2020-12-25 03:52:05.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/99/768499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/99/768499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/768499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/768499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/768499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/99/768499/1"}}},"commit":{"parents":[{"commit":"0bac3700b39424eced6aeb5419734855eeaf4462","subject":"Merge \"Update requirements to implement secure RBAC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0bac3700b39424eced6aeb5419734855eeaf4462"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 03:51:09.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 03:52:04.000000000","tz":-360},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: Ia0fa4ba1fecf2740a291c5a65b906b089c95404d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/552a20572f19ed0f60de863700583c722d7673a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/552a20572f19ed0f60de863700583c722d7673a0"}]},"branch":"refs/heads/master"},"a20dd92874ed4752d3bf1d73ce27239be7588323":{"kind":"REWORK","_number":2,"created":"2020-12-25 15:40:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/99/768499/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/99/768499/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/768499/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/768499/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/768499/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/99/768499/2"}}},"commit":{"parents":[{"commit":"0bac3700b39424eced6aeb5419734855eeaf4462","subject":"Merge \"Update requirements to implement secure RBAC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0bac3700b39424eced6aeb5419734855eeaf4462"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 03:51:09.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-25 15:40:32.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: Ia0fa4ba1fecf2740a291c5a65b906b089c95404d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a20dd92874ed4752d3bf1d73ce27239be7588323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a20dd92874ed4752d3bf1d73ce27239be7588323"}]},"branch":"refs/heads/master"},"5e7a0519cfc246dd2b5ca9a9749428cf0e5217e8":{"kind":"REWORK","_number":3,"created":"2020-12-25 16:41:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/99/768499/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/99/768499/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/768499/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/768499/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/99/768499/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/99/768499/3"}}},"commit":{"parents":[{"commit":"0bac3700b39424eced6aeb5419734855eeaf4462","subject":"Merge \"Update requirements to implement secure RBAC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/0bac3700b39424eced6aeb5419734855eeaf4462"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-12-25 03:51:09.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2020-12-25 16:41:53.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to do two things:\n\n1. Change the default value of \u0027[oslo_policy] policy_file\u0027\u0027\nconfig option from \u0027policy.json\u0027 to \u0027policy.yaml\u0027 with\nupgrade checks.\n\n2. Deprecate the JSON formatted policy file on the project side\nvia warning in doc and releasenotes.\n\nAlso replace policy.json to policy.yaml ref from doc and tests.\n\n[1]https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: Ia0fa4ba1fecf2740a291c5a65b906b089c95404d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5e7a0519cfc246dd2b5ca9a9749428cf0e5217e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/5e7a0519cfc246dd2b5ca9a9749428cf0e5217e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
