)]}'
{"id":"openstack%2Faodh~778780","triplet_id":"openstack%2Faodh~master~I2fd374514fd6bf3be02191b7a23b50b6dec93e08","project":"openstack/aodh","branch":"master","topic":"python38","hashtags":[],"change_id":"I2fd374514fd6bf3be02191b7a23b50b6dec93e08","subject":"Use python 3.8 for tox","status":"MERGED","created":"2021-03-04 21:34:17.000000000","updated":"2021-04-07 06:00:35.000000000","submitted":"2021-04-07 05:59:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778780-python38","meta_rev_id":"9f1631260aff36c21c4aeca189f6dac1e2a52a34","_number":778780,"virtual_id_number":778780,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 05:59:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2021-03-30 13:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2021-04-07 05:48:36.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2021-04-07 05:48:36.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 21:50:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 05:48:36.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"f60e34f5090804be8237036fd44ea4fb1569504d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-04 21:34:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44b6bbea0543bb602cc52f2445e769ac61cd5fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 21:50:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293d3102b80e447dabc83c7a3cccb2e7 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84041966106d423e9bbe5a7588faa7a9 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cce340e0e57542bababbd3842417ec23 : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50735cbe0fda41d1b8bad656855dd744 : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fce06204d5b41dcaf5d78214617556e : SUCCESS in 14m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a28c6af11f6f4e40abc3c7a370527820 : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":1},{"id":"eae09d4f6dc974b0bd020dd6b1975f35ccef0f40","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-30 13:47:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"74e677538512e0e9eee27e14dfcd93c9e74d9122","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-04-01 13:26:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bbaae7e49cb0513aa6d65671d2107f404c8952e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:47:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec349c78fbb4573b58bf1e2402072fe : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0439525f7502421b816043211605f6f9 : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6773d257be7a42feb46ddfc55b47cca9 : SUCCESS in 17m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/082b9d44e89247298b99aefa333512be : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63edc4d151e146fe90726a9b2fb7272d : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6beffb26b4a4a6d9d12de556d454374 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d16a9374ba757b679b7add293870817abffd7a5c","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-04-07 05:48:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f1b5d3eb693b6386f896d12373e64e381b0e5e31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 05:48:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c670681f1af7ba7d527bd440621104e9bc2aee7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 05:59:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/125a959ccc1d48298c39e7fe5125ce7d : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/562dc86ccf05434583205ef8e3fa6fee : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d1414a92c1d42628e0da82ec1e2a5c7 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82ac6e30c2940c0946f39ad1a7a43b0 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b592e2ed18b749e081325384b84569ac : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1ae29f27f236ade53bbe9934d2841668e688b179","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 05:59:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9f1631260aff36c21c4aeca189f6dac1e2a52a34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 06:00:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51eb7bd9df7944e2810d3a651a97d1eb : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd66b7fd64064daaa6d2233117255c85 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d587a9e56c4070bb9dc02056090534f0f0c996f1","revisions":{"d587a9e56c4070bb9dc02056090534f0f0c996f1":{"kind":"REWORK","_number":1,"created":"2021-03-04 21:34:17.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/80/778780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/80/778780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/80/778780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/80/778780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/80/778780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/80/778780/1"}}},"commit":{"parents":[{"commit":"418d17c34baaf9081088ddb0ab5c97fba67ed156","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/418d17c34baaf9081088ddb0ab5c97fba67ed156"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-04 21:33:34.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-04 21:33:34.000000000","tz":60},"subject":"Use python 3.8 for tox","message":"Use python 3.8 for tox\n\nChange-Id: I2fd374514fd6bf3be02191b7a23b50b6dec93e08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d587a9e56c4070bb9dc02056090534f0f0c996f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/d587a9e56c4070bb9dc02056090534f0f0c996f1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
