)]}'
{"id":"openstack%2Faodh~782106","triplet_id":"openstack%2Faodh~master~I816613b861a084fb92ba2e6e4f72e7830f11c1e5","project":"openstack/aodh","branch":"master","topic":"add-xena-python-jobtemplates","hashtags":[],"change_id":"I816613b861a084fb92ba2e6e4f72e7830f11c1e5","subject":"Add Python3 xena unit tests","status":"MERGED","created":"2021-03-22 09:28:47.000000000","updated":"2021-08-31 01:56:28.000000000","submitted":"2021-08-31 01:51:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782106-add-xena-python-jobtemplates","meta_rev_id":"2c109b302894464ec45f7ac09b9a0c20ad8eb00a","_number":782106,"virtual_id_number":782106,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-31 01:51:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2021-03-22 10:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2021-08-31 00:45:16.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-23 01:03:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2021-08-31 00:45:16.000000000","_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 10:45:11.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2021-03-22 10:46:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-23 01:03:05.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2021-08-31 00:45:16.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"}],"messages":[{"id":"8567efac196b227de9ea365cee0a9d75ad8fee9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-22 09:28:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d9dc15b6545e5db8e29056e4cbeef2d67e0b8b4","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-22 10:45:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e394153b3f38cdf6b6138ce15242977aa1e823a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:46:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb03dd88c60943218193a424f554ac05 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7393b63e9384c84992363253f5d5257 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fe4ea984cc843149c8a04200d2d8860 : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/995466fc792c418aa409ac4030e405cd : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92bcf8deeeda4d7683e7e5fbbff12fd2 : SUCCESS in 9m 18s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/a9deca7a0316409da63e6968d762d591 : SUCCESS in 49m 14s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/cc84c840c2ed4707bd168ad55c56c6c2 : SUCCESS in 50m 22s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/830e3bbf026e451cb1d4f6cf262c741b : FAILURE in 1h 04m 38s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/aaa96e81fc2142469f05a4adfd72a094 : FAILURE in 1h 07m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6b52fdb88a3b7f61d9989d71f37bf72fb3f21a86","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-22 14:55:08.000000000","message":"Patch Set 1:\n\nI suspect 8gb of ram is a bit too little for running ausotscaling use cases.","accounts_in_message":[],"_revision_number":1},{"id":"8d23bcc42721feb6667a4cd1080a9c7dc97688e5","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-03-23 01:03:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e2b8359a3e932ae2aeeb3f262a6fbe4b293e51df","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-26 18:47:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"df3100602d5adf76302265ab27dbd4deab930d49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 20:03:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d2fa0d9bd6a4ac8bc2b1b89aec84d55 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efa7b46c51f94303bf6eac8cf7fe7cfe : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e7eb9b25224cbebac42d491ceba46c : SUCCESS in 10m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf36828183d840cbb891f78c2b98f8ae : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1af10b189ec642c985e1aaa7d36761c4 : SUCCESS in 8m 43s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/c514f40693a74dfb81a04c4dbb6717bd : SUCCESS in 51m 44s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/12ebd9cad69946cb982e407a49042711 : SUCCESS in 49m 33s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e23fc9a3ab3940aea4e0eb5bfe94c794 : FAILURE in 1h 08m 12s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/061748be970d4a7584289cf5d1ed6daf : FAILURE in 1h 06m 56s","accounts_in_message":[],"_revision_number":1},{"id":"867845913541cf9e3bf0420dd191062dea657f3d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-14 07:14:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f901996bf77e384c72f629390fd195161ea56f5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 09:31:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3d89fb3533a4e9c9d7e88e89b915f60 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dd02f0a3a8e431982c04748df995c00 : TIMED_OUT in 41m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c40b623fea64428abfd3b1308596de7 : TIMED_OUT in 41m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/add8f56b1ab34cb9b131afff4833af9c : TIMED_OUT in 41m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f32fec7c8db427293944c8000bc1832 : SUCCESS in 8m 15s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/105d6ba8c3a24af1af7e99ac1d1abe5d : SUCCESS in 42m 05s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/c5fdbd3aa94147fea9578a4e93a1a218 : TIMED_OUT in 2h 13m 19s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/cd754bc885534b83873cb609bf4351f3 : FAILURE in 54m 50s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/25963201f9ab4512a12de82ba7b67e38 : FAILURE in 55m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3243663d0472d9c3f31016062f90329933fdb0cd","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-15 12:13:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ff31e370df39805108f160db2e2b42fd9abac05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 14:31:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df42497a482e44c58c5a4f974f453906 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a9f5fa36b864ad79f569631cdfde379 : FAILURE in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/765f609bb4dc4e758d188c8f2921644e : TIMED_OUT in 41m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e714ec1d35644b7a138e8904a40c024 : TIMED_OUT in 41m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5ec3f78224c4f76874cbff1caf71e79 : SUCCESS in 9m 00s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/cfd4ea9c57be441cb13768181101eaf9 : SUCCESS in 49m 28s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4f616e8e5b254e2789941d85c207d7ea : TIMED_OUT in 2h 14m 36s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/10b0887a3de44569bb691ee6a2b26aed : FAILURE in 57m 44s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/79d7cbac57134d6ea5bbe24bedbab959 : FAILURE in 56m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f24945d8ce304367e6615cca53ac47bd13a974d","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-30 09:24:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1706b3f16e3709976700e5cf6091c94b0db7173c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 11:41:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/439ff9974eb145bea0e825c55fc9ecd0 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cceee92289b04e51a1503f8570e87f8e : TIMED_OUT in 42m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43704f8b432f4b82bbca7ddeb6b84123 : TIMED_OUT in 43m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04f9f233f28b4a1c9aa62aeb0c1ae100 : TIMED_OUT in 43m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18f584f9077b47909d557f436190ce8b : SUCCESS in 9m 43s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2f1128b4564e4016970d6c38482b5009 : FAILURE in 44m 18s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/21d0599ee2714b5dacd79ebef4f9b514 : TIMED_OUT in 2h 15m 37s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/c73e7d41ff13401ea8fbe8295816bc7c : FAILURE in 48m 05s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/ea6965ed2f3b47caa4e1007e5f746909 : FAILURE in 44m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5ee44aa8050488025a4e3ada562258438c40dc3","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-08-28 16:15:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b80cf3a789200382a772376256a559fb5a38d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-28 17:09:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ffec26e6af342ae81876e13e9a21687 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ddd610a45d34fc0a5f4e086e41c2d05 : SUCCESS in 8m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86dbd61d14ca4e988c755913b1b32b25 : SUCCESS in 9m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45332de86ccc498aac0d73003b685b14 : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f23af7cd02d4f3495722fe3c66a1f42 : SUCCESS in 8m 52s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/04b7ad06bdc24ae7a79ceed4d44fb0bc : SUCCESS in 39m 05s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/bbea84b5203c4a7b8074cf300b6a060b : SUCCESS in 38m 27s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/9c1151b0d6d14ec8ad1d17b3e36a3956 : SUCCESS in 42m 15s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/09e84224355343ce98a5c90b47c16929 : SUCCESS in 44m 16s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/2bf10e329e5741d6bdd5bd95b04493cd : SUCCESS in 48m 36s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/394350cb213148189797de7d8978c96b : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9377450046d41799fccc6402497333010e7a6609","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2021-08-31 00:45:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"007da7eb19885bfcb90e2ef7009a9b2150db13ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 00:45:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15e96f967a7ea29e79052003bfa2bc6aefb06ed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 01:51:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2006185d18bf46d6a36f9cce654959c2 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2edce1c5abd64689b15c1449ffaf889a : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de669750798a41469e1f35605c953180 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102c640f7eba4c5ab3763deaea3e10c0 : SUCCESS in 8m 30s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/749612c34b9c455fa7337bd87f9c832b : SUCCESS in 40m 21s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/69c5f08020974f2e9fe6d99d886923f7 : SUCCESS in 43m 30s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1cd9317d83cf4650882e75e88133000c : SUCCESS in 44m 00s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/ab8a36624c3441ed8954092c3f6c6dab : SUCCESS in 46m 51s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/63ded0baa15f4fbe95e06cc9bf4010e1 : SUCCESS in 53m 09s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/cb6136e2570a4a8d9c63edf534f28c92 : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1d6f49435e724048a49109bbcd483f62035e8f6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 01:51:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c109b302894464ec45f7ac09b9a0c20ad8eb00a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 01:56:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f466f114b69943d7801ff56b7f3f1e9f : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f483f6c47af1cd850152a71c62954ebe72c5b1e0","revisions":{"f483f6c47af1cd850152a71c62954ebe72c5b1e0":{"kind":"REWORK","_number":1,"created":"2021-03-22 09:28:47.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/06/782106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/06/782106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/782106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/782106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/06/782106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/06/782106/1"}}},"commit":{"parents":[{"commit":"6fc880f6b31cf27fecb7a976b078e39a147019cd","subject":"Update master for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/6fc880f6b31cf27fecb7a976b078e39a147019cd"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-22 09:28:45.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-22 09:28:45.000000000","tz":0},"subject":"Add Python3 xena unit tests","message":"Add Python3 xena unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for xena.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: I816613b861a084fb92ba2e6e4f72e7830f11c1e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f483f6c47af1cd850152a71c62954ebe72c5b1e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f483f6c47af1cd850152a71c62954ebe72c5b1e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
