)]}'
{"id":"openstack%2Faodh~792580","triplet_id":"openstack%2Faodh~master~I973eb15433e66c75366666940a841e5a579bf0e8","project":"openstack/aodh","branch":"master","topic":"bug/1929178","hashtags":[],"change_id":"I973eb15433e66c75366666940a841e5a579bf0e8","subject":"Deprecate unused [coordination] check_watchers","status":"MERGED","created":"2021-05-21 10:51:50.000000000","updated":"2021-06-02 17:43:13.000000000","submitted":"2021-06-02 17:11:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792580-bug/1929178","meta_rev_id":"48725291904c05b1401f9b8d8974bcfd9ef4640f","_number":792580,"virtual_id_number":792580,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-02 17:11:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2021-05-21 12:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2021-06-02 15:57:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 12:05:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 12:41:23.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"04a5c8997c195671718e4405eaca39971ee27681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-21 10:51:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f30588ca2bffe2798e5519c8f735722988a5edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:05:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d153e038bc4e10ae276e632f8eb6b1 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea8cda98fd174cb5bd2567d2015929ad : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43d7a5e5871241f6b801ea274c7ca8cf : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d18d8225d97426aad5d0ff75b5e9fb5 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dda7a4f6f1f84fbda2171ef9061b9dae : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa6e59aaa8974fcc87b18f7dbad3a662 : SUCCESS in 6m 26s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/effb7932be6d4b89be840a1fcfc1343c : SUCCESS in 49m 26s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/1b2bf5f10973459aba8a29a6dd490219 : SUCCESS in 49m 09s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/b968f478ed974bb1aabf873be34a2f38 : FAILURE in 58m 42s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/9275cf430acc4ca298fe08ca56e014a7 : FAILURE in 1h 05m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d620f8409f707c4e28b4bb51dd647a2943cf589","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-21 12:41:23.000000000","message":"Patch Set 1: Code-Review+2\n\nThank you, makes sense to remove this.","accounts_in_message":[],"_revision_number":1},{"id":"962f1415fd201eeeddedd2184c56bfecf25b7fd0","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-02 15:57:44.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"07d99f30b09493ca7131847306602be2854e4b3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 16:01:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78820ad6da83b193b37aa04de276393bbf3e9d93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 17:11:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fdf502102d9477a9a39c51a075cce01 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/635658d2d188433a818ca4b51f66797e : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e86e122389ac474a8397e5b52f4f7738 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338d8e5769664e14a08e95e1fa4825b7 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5bdf63f75b3441a829b7b79a811cd67 : SUCCESS in 7m 48s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2756cbc975c349fb99e7558f790836f0 : SUCCESS in 43m 31s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d69a737b4c1d4279a544f7dc39cca1bc : SUCCESS in 42m 50s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/96fa3210c81347068646dcbe0e217537 : FAILURE in 1h 01m 45s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/53a505560d2a40f185873ebacbb1249d : FAILURE in 57m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"973675da123cf928ba50cd1a780769d494a5cc8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 17:11:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"48725291904c05b1401f9b8d8974bcfd9ef4640f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 17:43:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/effcbaee7d8444cbaf21ad17a35b5f86 : SUCCESS in 2m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d201c3ceb4d54631b88a7f1e5a0f5b7c : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"082d67f5965c69ee72be4bb1e1c44fd9ff5526d5","revisions":{"082d67f5965c69ee72be4bb1e1c44fd9ff5526d5":{"kind":"REWORK","_number":1,"created":"2021-05-21 10:51:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/792580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/80/792580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/80/792580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/80/792580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/80/792580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/80/792580/1"}}},"commit":{"parents":[{"commit":"352a6b4b42f50930ec188b4a6b71b55ccc9094fc","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/352a6b4b42f50930ec188b4a6b71b55ccc9094fc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-21 10:50:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-21 10:51:34.000000000","tz":540},"subject":"Deprecate unused [coordination] check_watchers","message":"Deprecate unused [coordination] check_watchers\n\nThe [coordination] check_watchers is no longer used in any\nimplementation. This change deprecates the parameter so that we can\nremove it in a future release.\n\nCloses-Bug: #1929178\nChange-Id: I973eb15433e66c75366666940a841e5a579bf0e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/082d67f5965c69ee72be4bb1e1c44fd9ff5526d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/082d67f5965c69ee72be4bb1e1c44fd9ff5526d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
