)]}'
{"id":"openstack%2Faodh~792581","triplet_id":"openstack%2Faodh~master~Ia86465eddb2b6b0d539628052f6e0a9ffbe17a3e","project":"openstack/aodh","branch":"master","topic":"bug/1929178","hashtags":[],"change_id":"Ia86465eddb2b6b0d539628052f6e0a9ffbe17a3e","subject":"DNM: Remove check_watchers","status":"ABANDONED","created":"2021-05-21 10:51:50.000000000","updated":"2021-06-26 17:54:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2802468f3223cf06687bf024801822aaff9bff46","_number":792581,"virtual_id_number":792581,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-21 11:59:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 11:59:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ac007f1a21c2dcd4ccbfe1e3ab84986c42f2c30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-21 10:51:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"783088e67015646dc15268469fb23df6058474ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 11:59:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a310d9e01084c15b21726699889fd0c : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c825c57d06b94bccbc75f0e3ebef652f : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/467f3e63b3964e8087ca1cf8ad50eaf5 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5dbf20adc80482ba17e7604e9312c66 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a5ae9436bae4b3a8d2fb1fe2f37af62 : SUCCESS in 9m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/ec820f362c7c481c8ce6c8bb2fea50f4 : SUCCESS in 41m 20s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/19b0b046d4254f569b4a538ffef3f554 : SUCCESS in 32m 14s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/01482cd4798e453eb108b7feb43d1c8f : POST_FAILURE in 1h 06m 14s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/de1874deef6044f78544134fcbdbc1e9 : FAILURE in 57m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2802468f3223cf06687bf024801822aaff9bff46","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-26 17:54:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fb2001495aeaaa4d4e680ee58a936451284a002","revisions":{"3fb2001495aeaaa4d4e680ee58a936451284a002":{"kind":"REWORK","_number":1,"created":"2021-05-21 10:51:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/81/792581/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/81/792581/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/81/792581/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/81/792581/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/81/792581/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/81/792581/1"}}},"commit":{"parents":[{"commit":"082d67f5965c69ee72be4bb1e1c44fd9ff5526d5","subject":"Deprecate unused [coordination] check_watchers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/082d67f5965c69ee72be4bb1e1c44fd9ff5526d5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-21 10:51:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-21 10:51:39.000000000","tz":540},"subject":"DNM: Remove check_watchers","message":"DNM: Remove check_watchers\n\nChange-Id: Ia86465eddb2b6b0d539628052f6e0a9ffbe17a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3fb2001495aeaaa4d4e680ee58a936451284a002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/3fb2001495aeaaa4d4e680ee58a936451284a002"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
