)]}'
{"id":"openstack%2Faodh~800070","triplet_id":"openstack%2Faodh~master~I9dc4e57a4bf4e70824b2b54b1a481c91267a29b6","project":"openstack/aodh","branch":"master","topic":"sqlalchemy","hashtags":[],"change_id":"I9dc4e57a4bf4e70824b2b54b1a481c91267a29b6","subject":"SQLAlchemy url is now immutable","status":"ABANDONED","created":"2021-07-08 15:19:08.000000000","updated":"2021-07-09 13:25:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8e1ee700c01c65a6ecbafdc1e35b81e577028b68","_number":800070,"virtual_id_number":800070,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-09 13:07:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-08 16:15:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-09 12:05:41.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"d3a22507ada2b3232b71d3a90ac11ed822980cdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-08 15:19:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fe4ed39b202df3949f5f1acba44524ad8c09c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 16:15:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fe06d34e3f9477f8feee498cccf44fa : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9195175122074833a411cf31a47c7349 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e912d7e16a5943a69fd2f7ea7e1669ae : SUCCESS in 8m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da6f7aa363b348ce917c35b622d40dac : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1180ffb762ca4912a7adf6a74a526d85 : SUCCESS in 8m 32s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/056309d1929a428cb716cb7fb4059eb9 : FAILURE in 35m 34s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/b17770193a3e4c879211c975f71d5ef7 : FAILURE in 47m 49s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/d80cc7d24ca4416caa2154c16e51887f : FAILURE in 48m 33s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d1178751d9ce4d9c91e5ee348e1acbb9 : FAILURE in 49m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7e6f2ef25d6cfcac663a9902409c3ad05a6150c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 05:35:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e443406c544a45e32216ce200022b53cd5ea28ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 06:29:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ed0464eb1f43ee947b58aa36d80441 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86622177215a4d88b71b7a7d0daff573 : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a823013a24b340f686cc9fcd1010e94a : SUCCESS in 9m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/273dc262f56e471fb3010cfc89b4d331 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b9be4e647b74ba4abb52c3ae0047998 : SUCCESS in 7m 45s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/5d39d872e85143be941368b92812065b : FAILURE in 40m 30s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/4245cd00bbf04be89f3c0d5a24de1b83 : FAILURE in 46m 57s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a78dc4c1b0764c53b5b3e81bde9e6fcf : FAILURE in 41m 37s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/4147d2a221c2469c91649a03aa4ef1b6 : FAILURE in 45m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58544a7dc5282c3a6ffe30787f2cf0f3ccf86c51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 09:53:19.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"73e0e53dbd0045d0f43ebb2cf4710f4c1d93b1b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 10:48:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b73d504864b4395bb7bfdc250c9e0d7 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cec9dc2730314f5e95d4e67674d74019 : SUCCESS in 8m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c04b9f322fd844f3864d9b784d16554f : SUCCESS in 8m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19c27e4bdaa446bbae870b01f3186fdf : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ed52229af1440c886ac4bb5b7f81cc : SUCCESS in 8m 25s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/af9e2206fa204219b1090673befb567d : SUCCESS in 51m 33s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/d05136aebd24451ba1932d150369d70c : SUCCESS in 36m 45s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1c21f506b99e4ce79bd5e803034b545c : FAILURE in 49m 40s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/51d572cb651e46a7ae242b104cdc3b76 : FAILURE in 53m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf48a4f94a4d80aa3f17e6b2547897a38263e852","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-09 12:05:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"96a3c3aaab6a72868a4573e6663ca1db0455ed85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-09 12:08:30.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"58ec7d3b64f8e113f0fac4f55f01eec0ed84d7a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 13:07:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6000930ac26041448c40d05fe2370814 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/466f4b18860d4cd4a562c1158cb6bc61 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/730c73ff279d47bc82b86d794f15e7fe : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/631aed26326f4771b7cc79190afe5dfc : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b31cc58162c4d268ee2cdc0b4094350 : SUCCESS in 7m 59s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6db2996f3c34466cbdd58841a6a9c09a : FAILURE in 41m 43s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/da66a3e0a4784ab2a1e6f709ea168b0c : FAILURE in 43m 57s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/bd9484fdd244437484193d2296cea49b : FAILURE in 45m 49s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/bbfd4fa1880241fe9eeba94d221ecb08 : FAILURE in 52m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8e1ee700c01c65a6ecbafdc1e35b81e577028b68","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 13:25:04.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.opendev.org/c/openstack/aodh/+/800198","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"032759b2144fba4e79fbed0a5171fcccf3b3d477","revisions":{"9bdb5b3f6ab9f8a77baee74c9e2fc126bf6b8597":{"kind":"REWORK","_number":1,"created":"2021-07-08 15:19:08.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/70/800070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/70/800070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/800070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/800070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/800070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/70/800070/1"}}},"commit":{"parents":[{"commit":"8404e9ac7407d440172f54ac83c7b11f84332d4a","subject":"Merge \"Bugs are in launchpad\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8404e9ac7407d440172f54ac83c7b11f84332d4a"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-08 15:09:07.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-08 15:18:51.000000000","tz":120},"subject":"SQLAlchemy url is now immutable","message":"SQLAlchemy url is now immutable\n\nUse URL.set to update\n\nChange-Id: I9dc4e57a4bf4e70824b2b54b1a481c91267a29b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9bdb5b3f6ab9f8a77baee74c9e2fc126bf6b8597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/9bdb5b3f6ab9f8a77baee74c9e2fc126bf6b8597"}]},"branch":"refs/heads/master"},"f064a1ae457681ac08c33724bd79f43380a2544a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-09 09:53:19.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/70/800070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/70/800070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/800070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/800070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/800070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/70/800070/2"}}},"commit":{"parents":[{"commit":"8404e9ac7407d440172f54ac83c7b11f84332d4a","subject":"Merge \"Bugs are in launchpad\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8404e9ac7407d440172f54ac83c7b11f84332d4a"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-08 15:09:07.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-09 09:53:11.000000000","tz":0},"subject":"SQLAlchemy url is now immutable","message":"SQLAlchemy url is now immutable\n\nUse URL.set to update\n\nDepends-on: https://review.opendev.org/c/openstack/aodh/+/799651/\nChange-Id: I9dc4e57a4bf4e70824b2b54b1a481c91267a29b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f064a1ae457681ac08c33724bd79f43380a2544a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/f064a1ae457681ac08c33724bd79f43380a2544a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"032759b2144fba4e79fbed0a5171fcccf3b3d477":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-09 12:08:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/70/800070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/70/800070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/800070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/800070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/70/800070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/70/800070/3"}}},"commit":{"parents":[{"commit":"8404e9ac7407d440172f54ac83c7b11f84332d4a","subject":"Merge \"Bugs are in launchpad\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8404e9ac7407d440172f54ac83c7b11f84332d4a"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-08 15:09:07.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-09 12:08:22.000000000","tz":0},"subject":"SQLAlchemy url is now immutable","message":"SQLAlchemy url is now immutable\n\nUse URL.set to update\n\nChange-Id: I9dc4e57a4bf4e70824b2b54b1a481c91267a29b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/032759b2144fba4e79fbed0a5171fcccf3b3d477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/032759b2144fba4e79fbed0a5171fcccf3b3d477"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
