)]}'
{"id":"openstack%2Faodh~800198","triplet_id":"openstack%2Faodh~master~I59c4d93a5914b8e14f7ef93bcaff8d1caaf5e75a","project":"openstack/aodh","branch":"master","topic":"gate","hashtags":[],"change_id":"I59c4d93a5914b8e14f7ef93bcaff8d1caaf5e75a","subject":"Replace oslo_utils.fnmatch with fnmatch","status":"MERGED","created":"2021-07-09 10:47:32.000000000","updated":"2021-07-09 14:18:05.000000000","submitted":"2021-07-09 14:16:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800198-gate","meta_rev_id":"7e66ec615f41d49151fbf470074500ee9c2a9135","_number":800198,"virtual_id_number":800198,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-09 14:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-07-09 13:11:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-07-09 13:20:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-07-09 13:20:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-09 12:03:21.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-07-09 12:33:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b8bc027bc3bcf5791967789b2970191447b4253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 10:47:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2161f4fce2f23306f404b3ee0c761e77ba6588b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-09 12:03:21.000000000","message":"Patch Set 1: Code-Review-1\n\nBecause these two changes fix different things I prefer merging these two separatedly.\n\nI rebased my change [1] on yours[2] . Because [1] passed CI so we can merge these two, one by one.\n\n[1] https://review.opendev.org/c/openstack/aodh/+/799651\n[2] https://review.opendev.org/c/openstack/aodh/+/800070","accounts_in_message":[],"_revision_number":1},{"id":"4d7202cecd93d3aa0abfba9450115bec5f9764b9","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 12:24:57.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Because these two changes fix different things I prefer merging these two separatedly.\n\u003e \n\u003e I rebased my change [1] on yours[2] . Because [1] passed CI so we can merge these two, one by one.\n\u003e \n\u003e [1] https://review.opendev.org/c/openstack/aodh/+/799651\n\u003e [2] https://review.opendev.org/c/openstack/aodh/+/800070\n\nMy patch depended on yours, that\u0027s why it passed finally.","accounts_in_message":[],"_revision_number":1},{"id":"41a6a03ce02f10e0aebe5e73947d64413b862ca7","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 12:29:55.000000000","message":"Patch Set 1:\n\nI general, I agree with you. Disconnected issues should go with separate patches.\nThis here is mostly due to updated dependencies, like sqlalchemy 1.4 and oslo_utils.\n\nIf we call this: fix telemetry-tempest, then it could go in one patch ;-)","accounts_in_message":[],"_revision_number":1},{"id":"2b290c8e7f5ee79a6c4191b7416cc15875093b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 12:33:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24ad87df4a124503a91fbe3f2f0509b4 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7aa4af27f7d4c30bfd7ccc9e1c02c6f : SUCCESS in 8m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f588708834f14be8886088b08b68354f : SUCCESS in 8m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc92fd58184840118e3c22f85f203673 : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd232298d0bf4b7683ed0cd0bdee24fc : SUCCESS in 7m 59s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/035f9b697960433ea9ac8dd69b4cfc9c : SUCCESS in 1h 11m 25s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/eac8afbf0c1c4a02beeba3b7051bdb9d : SUCCESS in 1h 23m 56s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/034e15a80ccf43c18ab9867327085a4d : FAILURE in 1h 34m 16s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/3d7d2016f9db421386a5201344e570f2 : FAILURE in 1h 39m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3efca8393f08ae33e99994a99dc9e462d8ac442d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-09 13:07:55.000000000","message":"Patch Set 1: -Code-Review\n\nAs we discussed in irc let\u0027s give another chance to https://review.opendev.org/c/openstack/aodh/+/800070 .\nIf it fails again I agree we merge this instead.","accounts_in_message":[],"_revision_number":1},{"id":"70aedd1269e25d728723d8bcbc6bc92bf0b8edb5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-09 13:11:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"351e1509d82bec0109e38bb7937b98116bdaf8cf","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-09 13:20:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you. I\u0027m going to merge this, it fixes the gating issues","accounts_in_message":[],"_revision_number":1},{"id":"5bd5d95705d1bc54b0a1ecc0d75c5cdb40c2972b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 13:21:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4e6fa2053033a62f2e1eaf50703a16377829445","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 14:16:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a4227d501404e549d0c20a12faac95b : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35e823f8bac349c1ac0f491224c1c337 : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b657d4c89e54eeaa7c40ff485bca846 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc979e895d6e4732be7f63304ce67444 : SUCCESS in 8m 07s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/0a6a2cdd47c241a3bf2c9a5cb5dd2f70 : SUCCESS in 43m 37s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/62e5fbea0fb5406eb787bcd1e71dbf6b : SUCCESS in 36m 16s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/27a4eefec7034c2bb883a38270caeca3 : FAILURE in 51m 33s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/fbf763f12ffa4ceba58867ee3094d388 : FAILURE in 54m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68c4f575c345f6d632edb919b2b32108e9fd4a0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 14:16:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7e66ec615f41d49151fbf470074500ee9c2a9135","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 14:18:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe6dd991c87740debce9915ffc231277 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a166e05f79817ddadea6fbf74a7d49972c05210","revisions":{"7a166e05f79817ddadea6fbf74a7d49972c05210":{"kind":"REWORK","_number":1,"created":"2021-07-09 10:47:32.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/98/800198/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/98/800198/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/800198/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/800198/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/98/800198/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/98/800198/1"}}},"commit":{"parents":[{"commit":"8404e9ac7407d440172f54ac83c7b11f84332d4a","subject":"Merge \"Bugs are in launchpad\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/8404e9ac7407d440172f54ac83c7b11f84332d4a"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-08 15:09:07.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-07-09 10:47:17.000000000","tz":120},"subject":"Replace oslo_utils.fnmatch with fnmatch","message":"Replace oslo_utils.fnmatch with fnmatch\n\nThe oslo_utils.fnmatch module was added to solve an issue in py2.7 but\nit is no longer required because py2.7 is no longer supported.\nThe module was deprecated since oslo.utils 4.9.1[1] and the stdlib\u0027s\nfnmatch module should be used instead.\n\nSQLAlchemy url is now immutable\nUse URL.set to update\n\nCo-authored by: Takashi Kajinami\n[1] 4c893c92f551c9dd2a7cfbe7ae8171ad8139df0b\n\nChange-Id: I59c4d93a5914b8e14f7ef93bcaff8d1caaf5e75a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7a166e05f79817ddadea6fbf74a7d49972c05210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7a166e05f79817ddadea6fbf74a7d49972c05210"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
