)]}'
{"id":"openstack%2Faodh~804291","triplet_id":"openstack%2Faodh~master~I4cfcb3e12012075eb56c6f69200825f0b597803f","project":"openstack/aodh","branch":"master","topic":"add_fips_job","hashtags":[],"change_id":"I4cfcb3e12012075eb56c6f69200825f0b597803f","subject":"Adding FIPS job","status":"MERGED","created":"2021-08-11 18:47:47.000000000","updated":"2021-08-24 16:51:59.000000000","submitted":"2021-08-24 16:47:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804291-add_fips_job","meta_rev_id":"7aecbeb77f2a0f9644f9fc4ac1d387c00325a11c","_number":804291,"virtual_id_number":804291,"owner":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 16:47:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":2,"date":"2021-08-18 15:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-18 14:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":1,"date":"2021-08-18 15:44:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 18:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-11 19:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-18 14:37:31.000000000","updated_by":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"reviewer":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"state":"REVIEWER"},{"updated":"2021-08-18 15:44:03.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"}],"messages":[{"id":"1e19eabecf5620760beef2e5fad51add8edc6627","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-11 18:47:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5147cc2498f1c024b63f1be11fd0c67a47234709","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-11 18:50:36.000000000","message":"Patch Set 1:\n\nBased this on what seemed to work for ceilometer. A lot of this is repetition that I think should actually move up to here: https://opendev.org/openstack/telemetry-tempest-plugin/src/branch/master/.zuul.yaml#L133","accounts_in_message":[],"_revision_number":1},{"id":"7336c52eacbabc622e3904417ce9895f7df2dc71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-11 18:50:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"45683e2580fefd1eaf1675311cf801f474bedae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 18:50:48.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"27579aec32bd887b138c69a4eacd3eb12f078c27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 19:46:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df6195e66b654752aef612c42b8d3fe4 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/499871b731e04ac08c249cf593ad6728 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b4da8d6a42b45f0a76b2d8b7d09cafa : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7d82b7acb1c4263bc02f3eaa7df4bd8 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f4c90f1437b4abab70324afd47334f8 : SUCCESS in 7m 34s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/95a5d9ae5f8f432282bb5a507d233dfa : SUCCESS in 46m 02s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/e53018f5555f4664b3696a30fb93c236 : SUCCESS in 42m 15s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/475fd15549b0449697e617899b4932e8 : SUCCESS in 49m 29s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/6526cc9a7c624d5da3c7d823e0616fdb : SUCCESS in 49m 44s (non-voting)\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/c7880d73eea240109acea5f042e37e02 : FAILURE in 29m 24s (non-voting)\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/914a754cd48248f9ba95ddca3d11e940 : FAILURE in 29m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4518963645cf9c799f7d33b7f15127137b880165","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-11 19:51:12.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"4ce7b048ba4fc1f32796c5b4e7b34c74db119e18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 19:51:51.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ceilometer on branch master.  The error was:\n\n  Job telemetry-dsvm-integration-centos-8s in openstack/ceilometer is not permitted to shadow job telemetry-dsvm-integration-centos-8s in openstack/aodh\n\nThe error appears in the following job stanza:\n\n  job:\n      name: telemetry-dsvm-integration-centos-8s\n      nodeset: devstack-single-node-centos-8-stream\n      parent: telemetry-dsvm-integration\n      description: |\n        Functional testing for a Centos 8 stream system\n\n  in \"openstack/ceilometer/.zuul.yaml@master\", line 28, column 3","accounts_in_message":[],"_revision_number":3},{"id":"0a77bea8cd2076669a2bf9779073a2b221b42c52","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-12 13:23:46.000000000","message":"Topic set to add_fips_job","accounts_in_message":[],"_revision_number":3},{"id":"ad753ca254057c57d8a235fe4a59c58e39902bd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-12 15:11:25.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d3f6719b2236f81c84c744f8253ac7809591c8dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 15:11:43.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":4},{"id":"8558bf4f3ee05f51f8444e4d2470c860b5bf964d","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-12 16:45:39.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2c64caf033dd91bb1a589cb3b357fd2c8e3a0b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 17:51:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a390f3fd54b1420e839dd369fc3046e2 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c69e74ca5114b47b6794118efb48bbd : SUCCESS in 8m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f9395c4f1c04010a85a907b8d06bacb : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47d7df5fc8f24537ac93880a4f052b98 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce14d2e22c8451bba97ed2cacf8a35d : SUCCESS in 7m 21s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/35f8d45df7f545c7972ee01b8b6fbe18 : FAILURE in 19m 27s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/37da1faa8a5a4b8e92f10b81ce81acf6 : FAILURE in 22m 01s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/8366e3840d4345048c771942e8d365ff : FAILURE in 22m 02s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/c2afdb2a65eb4b4eaf5fdb542c77aa5d : FAILURE in 25m 26s (non-voting)\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/8357f0c2a7524f3fb76dc8e4ed01e71c : SUCCESS in 50m 43s (non-voting)\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/5d754db52edc46c1ac0cc1d1baac181d : SUCCESS in 1h 00m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e6453eb389a336e12d14e835e64550c3aa572872","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-17 18:56:52.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"55e94df7eef9d2534f8221e389f385b0b974792b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 20:02:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841bdb290c8d4d5d8d847e2142f226fd : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7774b545604b47a0ac98aa7d526832e0 : SUCCESS in 9m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0691d68ae58e486f8b303a3ffc989b40 : SUCCESS in 8m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/634f0635ad374faabb33b38612174292 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf39d670acf4d25acce05d21634f0f2 : SUCCESS in 7m 53s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/36340d23c37f4b5e998e04829eabbe92 : SUCCESS in 43m 49s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/0c878fbc15cc45c292d352b7fe90bb9c : SUCCESS in 42m 02s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/1fd38c3ddcb746ddb9304725cd9c925e : SUCCESS in 48m 01s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/03329e3cd8b0414a9ba337e6707eadaa : SUCCESS in 48m 58s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/448c225630d34e1a9c1b4d57d6047c4b : SUCCESS in 53m 42s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/3118a562669341ed8e4e4f22d837d534 : SUCCESS in 58m 15s","accounts_in_message":[],"_revision_number":4},{"id":"58d1b773e3fb8423519a9ddfa732a460a34f0108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-18 04:31:59.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"b1c0f05fb2ea96b6292ba12f3649de7abc848be5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 05:25:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d289237335d5427b92c5d88a40ee0f8e : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ce25d4f2c0b452eac0299aa98596c19 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb3bdbcb6d0d4d4586a3bee0df08041e : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3023be1c95f84f9ba06d6286df6acc8a : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e13f302fae746069b3fea2b6ecac3fd : SUCCESS in 7m 33s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/22b1167d0717484a9a8f7b8f5c822824 : SUCCESS in 33m 52s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/f0dacbb7f6e144bc8f35b96602e299ef : SUCCESS in 39m 29s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/f3434bf3210c4550b1e3f847bbc8e702 : SUCCESS in 50m 33s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/49f4ac04b8a54e16bd66a06282c24a8e : SUCCESS in 45m 39s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/e37e855d83334eb8b58d653a2f0bef85 : SUCCESS in 42m 49s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/a0acefe20f1b4dc5af50478f8436e0b3 : SUCCESS in 46m 42s","accounts_in_message":[],"_revision_number":5},{"id":"4781322668ecd57362f78f4f371a969766c40159","author":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"date":"2021-08-18 14:37:31.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e0625a1c2b4cc499511f9679087f997f4bfc7737","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-08-18 15:44:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nlooks good to me, thanks!","accounts_in_message":[],"_revision_number":5},{"id":"b6a8b5083a770cf766faaeac44c64657aa5fbe17","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-19 18:27:02.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4e814ba629a9ab10ff1191c836afa5d3f8ed3111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 19:30:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d9340dc539c4b378c3bf2d182293ae1 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/115f9cae310a4e2a99cba083640633ba : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2975652568a54fbbb7f4d3111f8619d4 : SUCCESS in 8m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1a047a2b1fd473ba6702f3f19b1a20e : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a53c1718e09403a8dc0fde1988b80d6 : SUCCESS in 7m 51s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/f37036b235c14406a3bed3a0bdfd9519 : SUCCESS in 41m 19s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/489ea5fc3b4b4942964bec79ef9f20e3 : SUCCESS in 45m 24s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/a33e79602a274779963476ff150a6f0f : FAILURE in 20m 59s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/580dc66f916940a3931aaa1cd3c7524d : SUCCESS in 55m 25s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/e275254e00914bbcab14149854d477ac : SUCCESS in 53m 17s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/67fc4c718a7e4714bb377267b2cf872f : SUCCESS in 49m 02s","accounts_in_message":[],"_revision_number":5},{"id":"b1f863bdc7454c06addfbe75c1a7110a4bccc61f","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-24 14:45:50.000000000","message":"Patch Set 5:\n\nFailure was \"dpkg: error processing package pcp (--configure):\"  seems totally unrelated. It\u0027s been a few days, I\u0027ll try again","accounts_in_message":[],"_revision_number":5},{"id":"f821be165a29fdd15595e7b3463502512425e7bb","author":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"date":"2021-08-24 14:45:54.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"21776a9d14e697f85f2357955504ffedb87a7c96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 15:40:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af646a731b6e47bb94a7cb55d237cb44 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c318e1472fc41c9bfd14106ad738645 : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed079fd372a04dcfad80bd4f20a5b337 : SUCCESS in 10m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3f6fa83c78841c9b503bf5a5cc8a1f5 : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1138b67f416e4673a990fc3b7f3931d3 : SUCCESS in 9m 34s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/4394ab952cae4d748af1fe885be41a7b : SUCCESS in 40m 31s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/ae8b92412b704299b72e74b1bf4fe77a : SUCCESS in 40m 07s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/e1780beff9514a94bab3c767a6635d5a : SUCCESS in 45m 40s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/2f90f4dbe02946108e667223b94eaca9 : SUCCESS in 48m 14s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/5378d60c20a4411685efc32b58570b8f : SUCCESS in 48m 16s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/358e8a9b5073440f91c36b071987862e : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":5},{"id":"d484ed34b923a603520a3bcfb908de75613abba0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 15:41:25.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0489dafb27edf28aa0ded383aaadd1c2ef3df5b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 16:47:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e724323ecc064fd0a8f37db49f6c57e0 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/224d4ec1b167400ea21d014f3585217f : SUCCESS in 9m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8015eddf8344b4da6f794f56dfe7af7 : SUCCESS in 8m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c9b908e6f44a27af4234f46f6828d1 : SUCCESS in 11m 22s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/6f7dc122401e4c4e973d8eaf78b1d249 : SUCCESS in 40m 23s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/323323e4974f42e598ae7121450dcc9c : SUCCESS in 38m 57s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/08b1031c9222442fb6d5166398b5c50b : SUCCESS in 38m 02s\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/05945694d0644cfe9450ed15653758ad : SUCCESS in 39m 07s\n- telemetry-dsvm-integration-centos-8s https://zuul.opendev.org/t/openstack/build/f341c68abf24422abdd6ffa589109890 : SUCCESS in 53m 11s\n- telemetry-dsvm-integration-centos-8s-fips https://zuul.opendev.org/t/openstack/build/b36a9b26666647f9bef7cdb061a07c89 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":5},{"id":"c3ce6058c8fa3f7b68997d5bc5c6b959c8156f37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 16:47:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"7aecbeb77f2a0f9644f9fc4ac1d387c00325a11c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 16:51:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74dcf302183f4a649dcbe914236426a6 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a236b07d3d75292ebbaabe905acfcb213a2705ab","revisions":{"e60f5f7931894aad248171b93cf96a3175acf0d1":{"kind":"REWORK","_number":1,"created":"2021-08-11 18:47:47.000000000","uploader":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"ref":"refs/changes/91/804291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/91/804291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/91/804291/1"}}},"commit":{"parents":[{"commit":"10383f4d5b899d2644968090e4572c47fe314695","subject":"Merge \"Fix oslo policy DeprecatedRule warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10383f4d5b899d2644968090e4572c47fe314695"}]}],"author":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-11 18:46:19.000000000","tz":-240},"committer":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-11 18:46:19.000000000","tz":-240},"subject":"WIP/DNM - Adding FIPS job","message":"WIP/DNM - Adding FIPS job\n\nChange-Id: I4cfcb3e12012075eb56c6f69200825f0b597803f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e60f5f7931894aad248171b93cf96a3175acf0d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/e60f5f7931894aad248171b93cf96a3175acf0d1"}]},"branch":"refs/heads/master"},"13534573dbc21d89c82182eddf09d03322e9a335":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-08-11 18:50:45.000000000","uploader":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"ref":"refs/changes/91/804291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/91/804291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/91/804291/2"}}},"commit":{"parents":[{"commit":"10383f4d5b899d2644968090e4572c47fe314695","subject":"Merge \"Fix oslo policy DeprecatedRule warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10383f4d5b899d2644968090e4572c47fe314695"}]}],"author":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-11 18:46:19.000000000","tz":-240},"committer":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-11 18:48:18.000000000","tz":0},"subject":"WIP/DNM - Adding FIPS job","message":"WIP/DNM - Adding FIPS job\n\nChange-Id: I4cfcb3e12012075eb56c6f69200825f0b597803f\nDepends-On: Ie5dc0932ba5e413d909bd235058222ff52e30147\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/13534573dbc21d89c82182eddf09d03322e9a335"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/13534573dbc21d89c82182eddf09d03322e9a335"}]},"branch":"refs/heads/master","description":"Edit commit message"},"03c597b63910e94fc3036e9839c84c0b390d6272":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-08-11 19:51:12.000000000","uploader":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"ref":"refs/changes/91/804291/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/91/804291/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/91/804291/3"}}},"commit":{"parents":[{"commit":"10383f4d5b899d2644968090e4572c47fe314695","subject":"Merge \"Fix oslo policy DeprecatedRule warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10383f4d5b899d2644968090e4572c47fe314695"}]}],"author":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-11 18:46:19.000000000","tz":-240},"committer":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-11 19:51:01.000000000","tz":0},"subject":"WIP/DNM - Adding FIPS job","message":"WIP/DNM - Adding FIPS job\n\nChange-Id: I4cfcb3e12012075eb56c6f69200825f0b597803f\nDepends-On: Ie5dc0932ba5e413d909bd235058222ff52e30147\nDepends-On: Ic158561b3f4d8ddcc092f4e8da1b3215f976fc98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/03c597b63910e94fc3036e9839c84c0b390d6272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/03c597b63910e94fc3036e9839c84c0b390d6272"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7d90275a1884565f84492dce32fb753d4c204420":{"kind":"REWORK","_number":4,"created":"2021-08-12 15:11:25.000000000","uploader":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"ref":"refs/changes/91/804291/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/91/804291/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/91/804291/4"}}},"commit":{"parents":[{"commit":"10383f4d5b899d2644968090e4572c47fe314695","subject":"Merge \"Fix oslo policy DeprecatedRule warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10383f4d5b899d2644968090e4572c47fe314695"}]}],"author":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-11 18:46:19.000000000","tz":-240},"committer":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-12 15:11:12.000000000","tz":0},"subject":"WIP/DNM - Adding FIPS job","message":"WIP/DNM - Adding FIPS job\n\nChange-Id: I4cfcb3e12012075eb56c6f69200825f0b597803f\nDepends-On: I05713ed77b79e583ad1073ff80ccd7a031e34c26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7d90275a1884565f84492dce32fb753d4c204420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/7d90275a1884565f84492dce32fb753d4c204420"}]},"branch":"refs/heads/master"},"a236b07d3d75292ebbaabe905acfcb213a2705ab":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-08-18 04:31:59.000000000","uploader":{"_account_id":30893,"name":"Chris Sibbitt","email":"csibbitt@redhat.com","username":"csibbitt"},"ref":"refs/changes/91/804291/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/91/804291/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/91/804291/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/91/804291/5"}}},"commit":{"parents":[{"commit":"10383f4d5b899d2644968090e4572c47fe314695","subject":"Merge \"Fix oslo policy DeprecatedRule warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/10383f4d5b899d2644968090e4572c47fe314695"}]}],"author":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-11 18:46:19.000000000","tz":-240},"committer":{"name":"Chris Sibbitt","email":"csibbitt@redhat.com","date":"2021-08-18 04:31:06.000000000","tz":0},"subject":"Adding FIPS job","message":"Adding FIPS job\n\nChange-Id: I4cfcb3e12012075eb56c6f69200825f0b597803f\nDepends-On: I05713ed77b79e583ad1073ff80ccd7a031e34c26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a236b07d3d75292ebbaabe905acfcb213a2705ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/a236b07d3d75292ebbaabe905acfcb213a2705ab"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
