)]}'
{"id":"openstack%2Faodh~804595","triplet_id":"openstack%2Faodh~master~I56236286267f871977163f721de4e89e7d829862","project":"openstack/aodh","branch":"master","topic":"dnm","hashtags":[],"change_id":"I56236286267f871977163f721de4e89e7d829862","subject":"Testing gate (2)","status":"ABANDONED","created":"2021-08-15 23:49:02.000000000","updated":"2021-08-16 10:23:50.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a32455fed0315b21252bea8b38edd65063592fbb","_number":804595,"virtual_id_number":804595,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-16 09:16:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-16 00:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ece9a5f4f38885037399ab79fa1ec75c253cec4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-15 23:49:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffe1cb6d6c36d05b4565ee59973ec107e68c8bc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 00:12:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a38845d672c84e9694ac23d8ca78c88d : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c26b1a23efb49959ee862313b796183 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74ef38c106fc4a84a018d7e2f1220c24 : SUCCESS in 8m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2673a4b338a43ea8139d442a71f6f55 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52b03072d2bc47a58475fbda04f3d91a : SUCCESS in 9m 10s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/92d2cd8351e74200aa8ba4d669dc117f : FAILURE in 13m 04s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/7ffbe8598b824411b60a21412dfe3d05 : FAILURE in 11m 24s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/32f64e5082a442b598fdd122310e7f30 : FAILURE in 11m 25s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/d7a3ed435ae7484c95de8ca3911d2e49 : FAILURE in 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d6c3d53e350cc70ddeb74cdc67305dc0610c8a60","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-16 08:30:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b2779fca29f8f3f3b48d507bc4bc73d160c20db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-16 09:16:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d16ae118aec46bb84fd2fce6dacba5c : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ae23e80aeca4626a75b0ed248c90e9a : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f26536d48a7475cbf2f54ddf4ba6232 : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d7af05c514849dc8a965fa74b459ea6 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e15b79b10f6d4d58b7ffa28b804b121a : SUCCESS in 10m 11s\n- aodh-tempest-plugin-mysql https://zuul.opendev.org/t/openstack/build/2391d8a96cd94bedbb4a3a9732603d7c : SUCCESS in 38m 36s\n- aodh-tempest-plugin-postgresql https://zuul.opendev.org/t/openstack/build/57b2689111a14a66a04891148aecda9d : SUCCESS in 42m 55s\n- telemetry-dsvm-integration https://zuul.opendev.org/t/openstack/build/7033a93c66e64ab08f545d04bd27035c : SUCCESS in 34m 15s (non-voting)\n- telemetry-dsvm-integration-ipv6-only https://zuul.opendev.org/t/openstack/build/37b9997988e84a689961b18130bf95f3 : SUCCESS in 45m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a32455fed0315b21252bea8b38edd65063592fbb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-16 10:23:50.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b3ab18ee31b2841f35ddb007c741cc29a85890d","revisions":{"4b3ab18ee31b2841f35ddb007c741cc29a85890d":{"kind":"REWORK","_number":1,"created":"2021-08-15 23:49:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/804595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/aodh","ref":"refs/changes/95/804595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/804595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/804595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/aodh refs/changes/95/804595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/aodh refs/changes/95/804595/1"}}},"commit":{"parents":[{"commit":"b3246536ec64437e68e1e05f1d2d2bd22f010194","subject":"DNM: Testing gate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/b3246536ec64437e68e1e05f1d2d2bd22f010194"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-15 23:47:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-15 23:47:40.000000000","tz":540},"subject":"Testing gate (2)","message":"Testing gate (2)\n\nDepends-on: https://review.opendev.org/c/openstack/ceilometer/+/804587\nChange-Id: I56236286267f871977163f721de4e89e7d829862\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4b3ab18ee31b2841f35ddb007c741cc29a85890d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/aodh/commit/4b3ab18ee31b2841f35ddb007c741cc29a85890d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
